)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~671577","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~I4aece72ccdbb957eef9d69cb32ec6ffff5fdf8e0","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I4aece72ccdbb957eef9d69cb32ec6ffff5fdf8e0","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:16:21.000000000","updated":"2019-07-19 12:45:19.000000000","submitted":"2019-07-19 12:45:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671577-1563540319280-a431606b","meta_rev_id":"4848b8e3f6f599f9ab3a2c50b9a3e0d468554e9e","_number":671577,"virtual_id_number":671577,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 12:45:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 12:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-19 12:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 12:23:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 12:45:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11d462972f0b6bf104c053953e1354d50c05da9c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ece12e86b4ba7c699b4d79358ce81f57ec94ccac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:48:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/671577/1/check/openstack-ansible-linters/3740b37/ : SUCCESS in 9m 39s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/77/671577/1/check/openstack-ansible-functional-debian-stable/bd0105b/ : SUCCESS in 9m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/77/671577/1/check/openstack-ansible-functional-centos-7/fd907b9/ : SUCCESS in 6m 28s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/77/671577/1/check/openstack-ansible-functional-ubuntu-bionic/853064a/ : SUCCESS in 5m 25s\n- openstack-tox-docs http://logs.openstack.org/77/671577/1/check/openstack-tox-docs/e8eb79a/html/ : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2ade6d425bcdc0d3bc6fe7c11c46b5fd0166c331","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 12:23:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebad06b18a91da7f5f28cf2c2ad557f485b198d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:24:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9779837eab1bb54c7d236a153fc8da2d94484234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:45:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"610887d1e4ae6b707576ed03561addba05d680aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:45:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/671577/1/gate/openstack-ansible-linters/47ba167/ : SUCCESS in 5m 33s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/77/671577/1/gate/openstack-ansible-functional-debian-stable/595fcc7/ : SUCCESS in 7m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/77/671577/1/gate/openstack-ansible-functional-centos-7/39bb7d2/ : SUCCESS in 7m 33s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/77/671577/1/gate/openstack-ansible-functional-ubuntu-bionic/36be2af/ : SUCCESS in 9m 13s\n- openstack-tox-docs http://logs.openstack.org/77/671577/1/gate/openstack-tox-docs/5b33d78/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22221481ffae46946eaed6dc7ff1fc03df85ccf0","revisions":{"22221481ffae46946eaed6dc7ff1fc03df85ccf0":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:16:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/671577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/77/671577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/671577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/671577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/671577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/671577/1"}}},"commit":{"parents":[{"commit":"54020a6c142298684e4cde8906f286130ef40f51","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/54020a6c142298684e4cde8906f286130ef40f51"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:16:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:16:19.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I4aece72ccdbb957eef9d69cb32ec6ffff5fdf8e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/22221481ffae46946eaed6dc7ff1fc03df85ccf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/22221481ffae46946eaed6dc7ff1fc03df85ccf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
