)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~719377","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~I92e5b02bccab1bbe525055ebf04754b0870d6fb4","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","topic":"dev","hashtags":[],"change_id":"I92e5b02bccab1bbe525055ebf04754b0870d6fb4","subject":"Missing document start \"---\"","status":"ABANDONED","created":"2020-04-13 02:37:55.000000000","updated":"2021-01-19 15:30:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c032da4af4694c6c358362f82fbd08861afdd67c","_number":719377,"virtual_id_number":719377,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-13 02:48:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-04-13 05:50:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 02:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-13 05:50:34.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"7f8ad27c451f0c328c32402538df0118acb1c7e7","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-04-13 02:37:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b64c6838bb1c7cbfa40e84597b36314df2d533ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 02:48:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/508b45a70684440d91f6a99ff214e472 : SUCCESS in 6m 29s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/beb46b5c9e3d4c50b10b5d16ee75575a : SUCCESS in 6m 19s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9737f31c76dc4dea86f4f88cd58f83b4 : SUCCESS in 7m 17s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf53005a50f742a4baa16bca1eb8301c : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf88fb2db3b458eaba035bcfe6c6bdc : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ab5ffa878aac3ecaee7659e01b68b71615506afb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 05:50:34.000000000","message":"Patch Set 1: Code-Review-1\n\nthis is not needed for Zuul, works just fine as is.","accounts_in_message":[],"_revision_number":1},{"id":"c8262024849bbf7377702a344c3946f9a74674a2","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-04-13 05:54:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"9ec9ca4f98049990acc456b9bd3a10a73e4f8d0c","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2020-04-13 05:56:51.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"0b478c96b267c19f3bd4c894afb8901d471570a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 06:09:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1a9c95437b044b418f449495fd25439e : SUCCESS in 10m 02s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2c86de436b484df6a3e42a35acc0c9a8 : SUCCESS in 7m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e61c3f769773473e87f93cc5b85e6535 : SUCCESS in 9m 53s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d9d02301b9c4872a08ca9268ea3b2c0 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea939152618476c81915eded3255e20 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c032da4af4694c6c358362f82fbd08861afdd67c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-19 15:30:44.000000000","message":"Abandoned\n\nthis change is not necessary","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c313007db87a43687ba79e6404aa073d85f1177f","revisions":{"c313007db87a43687ba79e6404aa073d85f1177f":{"kind":"REWORK","_number":1,"created":"2020-04-13 02:37:55.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/77/719377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/77/719377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/719377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/719377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/719377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/77/719377/1"}}},"commit":{"parents":[{"commit":"9f75bc2cd31640646488e1875fa0eadde00211df","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/9f75bc2cd31640646488e1875fa0eadde00211df"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2020-04-13 02:36:49.000000000","tz":480},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2020-04-13 02:37:18.000000000","tz":480},"subject":"Missing document start \"---\"","message":"Missing document start \"---\"\n\nChange-Id: I92e5b02bccab1bbe525055ebf04754b0870d6fb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/c313007db87a43687ba79e6404aa073d85f1177f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/c313007db87a43687ba79e6404aa073d85f1177f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
