)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~727289","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~Iad7064a7f466cc97693818786333e61fad558a89","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iad7064a7f466cc97693818786333e61fad558a89","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-05-12 17:48:46.000000000","updated":"2020-05-12 19:19:15.000000000","submitted":"2020-05-12 19:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727289-1589310962392-793a8d12","meta_rev_id":"632c1bc11cbe36c3be28752d8e5af836ed3cc826","_number":727289,"virtual_id_number":727289,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 19:16:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-12 19:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-12 19:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 19:01:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-12 19:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2817c195ca21129b62e9754af853fce72d9973a8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 17:48:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"116824940c8396a82a57b061ef1865005b01f932","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 18:02:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3ecea0e2b012f44db4213b9d8b1f945acb4403c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:20:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7ad943f1a212475798f6ed89de8bac08 : SUCCESS in 9m 42s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/89bafa6a432443599968c32684e6f9d7 : SUCCESS in 8m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e91b89e52744439d9ed0137042913ae7 : SUCCESS in 10m 41s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0696d7048f3b4120b83ca85280b30bd8 : SUCCESS in 7m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/012a060bce91429c8194fd06c94dd470 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a991bda76ea44e29659e438ad68561b : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa2b05ce7a024653a9b57c5fd71a8676 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e813f60eca195973ab389406f23bd2022a794f4d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 19:01:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"819d0050149efcfbdf1dfd44dc04e825f9de3998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:04:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"484723053bb2d71c1762d97203b9c3995dc65103","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:16:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b8ea232d67a73bb8888a0852eda5879e9ae20917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:16:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/50d148453241402180d68bf8e18ed6fa : SUCCESS in 7m 20s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c6953f5e08cd40418d5bde82512e32e3 : SUCCESS in 6m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5adeb804f12f4345a2303bb8253deaa5 : SUCCESS in 6m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c753bc94ec13424ab517fadf4075ee05 : SUCCESS in 7m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b02efe3ab7849b5a2a152f101e91d1f : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3af572e4f59465aa6b9da5c355982c5 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e4ada0037c467cbb84a5d180c7fe3b : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"caeddbd4ee9a89c9ddc2672c84663d19f98fe56f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:19:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11324fbd1e794a7ca92f4696a293f5a9 : SUCCESS in 2m 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a2b42dd44f440d8305817b7a93a9ba : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1","revisions":{"2b544ce789fa15fb10ec0c16da9d7f1b87a0ed2b":{"kind":"REWORK","_number":1,"created":"2020-05-12 17:48:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/727289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/89/727289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/1"}}},"commit":{"parents":[{"commit":"14209056a181cc2442909f90a59a745cb1123bb3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/14209056a181cc2442909f90a59a745cb1123bb3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:48:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:48:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iad7064a7f466cc97693818786333e61fad558a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/2b544ce789fa15fb10ec0c16da9d7f1b87a0ed2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/2b544ce789fa15fb10ec0c16da9d7f1b87a0ed2b"}]},"branch":"refs/heads/master"},"4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1":{"kind":"REWORK","_number":2,"created":"2020-05-12 18:02:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/727289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/89/727289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/89/727289/2"}}},"commit":{"parents":[{"commit":"14209056a181cc2442909f90a59a745cb1123bb3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/14209056a181cc2442909f90a59a745cb1123bb3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:02:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:02:41.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iad7064a7f466cc97693818786333e61fad558a89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
