)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~729668","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~If62052b713e054d70fe389529dae02ff771fd43d","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","topic":"pygments_style","hashtags":[],"change_id":"If62052b713e054d70fe389529dae02ff771fd43d","subject":"Update docstheme for style","status":"MERGED","created":"2020-05-20 16:44:09.000000000","updated":"2020-05-22 09:42:48.000000000","submitted":"2020-05-22 09:39:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729668-1590140379691-fdce670a","meta_rev_id":"ff051c01aaa0b04cfc3f26d53a89261ad5122803","_number":729668,"virtual_id_number":729668,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 09:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-05-22 09:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-20 20:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-05-22 09:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 16:46:56.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-05-20 20:32:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-22 09:23:57.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-05-22 09:39:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1977f8842c658649f9626793056fcb16be8befa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 16:44:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29a0d3a60f35ddee983ce87c05c59828370dc420","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-05-20 16:46:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"447bcb28844e622d25a5b44043618d812148dff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:03:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/89a1f4aaa1304282a701bb64e52d0710 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fdb6947a2c74503a5d2ce995d7f1c82 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebae53198d184368b4dd3e17057bdadd : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"185eebd02749bf45e5e6e0fa1b98fcfd5c2a5462","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 17:24:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c10dda6fefe96d9f92371de84eee4d53c9ee66bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:40:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/713a416f91f346a79da563b966cca669 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63fde54559a475796210ebf0437ab8d : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa8f1d68bae44feeab93a907b627a4b4 : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6f9ffdae5c160463027f0a056bb0ffbbfd6779e4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 20:32:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8acd5f72070de024ab5002ebb16702e4f1caf991","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-22 09:23:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0686f5ab27656828e4ffc401dd7571cde1712482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:24:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8231acc5b6572df119a74f1880dd75cf4d7000f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:39:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6578c41e8ea14a508d9d127adcf7582e : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec27e18e4c604f9db72514207bf31914 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad661d0e203e4980b244f940a14d6ebb : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"22fb066f11ca1cddab6a94012e6bfaaeeab97dd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:39:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2c5387bb46f9ec8b253bdec2d56885374eebca97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 09:42:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e766b16fe214ea396c7f992ff26cb22 : SUCCESS in 2m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3ae1165dbac4e5e9db3603a4a700456 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05577fbb0bed833b78226c489f4f5da8f07b9b3b","revisions":{"6a55ca913d801c2909bd11982aeb183429fbaa55":{"kind":"REWORK","_number":1,"created":"2020-05-20 16:44:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/68/729668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/68/729668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/1"}}},"commit":{"parents":[{"commit":"4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:44:04.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:44:04.000000000","tz":120},"subject":"Update docstheme for style and PDF","message":"Update docstheme for style and PDF\n\nNew version of openstackdocstheme (Victoria+) respects pygments_style.\nSince this repo is using now Victoria (master) requirements but has\nnot branched for Ussuri yet, it uses the new version.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nChange-Id: If62052b713e054d70fe389529dae02ff771fd43d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/6a55ca913d801c2909bd11982aeb183429fbaa55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/6a55ca913d801c2909bd11982aeb183429fbaa55"}]},"branch":"refs/heads/master"},"05577fbb0bed833b78226c489f4f5da8f07b9b3b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-20 17:24:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/68/729668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/68/729668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/68/729668/2"}}},"commit":{"parents":[{"commit":"4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/4e85e7ff0d4c3bf7dc364bbc718a67841b0449d1"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:44:04.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-20 17:24:09.000000000","tz":0},"subject":"Update docstheme for style","message":"Update docstheme for style\n\nNew version of openstackdocstheme (Victoria+) respects pygments_style.\nSince this repo is using now Victoria (master) requirements but has\nnot branched for Ussuri yet, it uses the new version.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nChange-Id: If62052b713e054d70fe389529dae02ff771fd43d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/05577fbb0bed833b78226c489f4f5da8f07b9b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/05577fbb0bed833b78226c489f4f5da8f07b9b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
