)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~762042","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~Ie74f009f2ef0c56acc49704902095789fb24445a","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","hashtags":[],"change_id":"Ie74f009f2ef0c56acc49704902095789fb24445a","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"MERGED","created":"2020-11-10 05:11:28.000000000","updated":"2020-11-10 12:01:19.000000000","submitted":"2020-11-10 11:59:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762042-1605009591911-e490bc4f","meta_rev_id":"741447bdf598da457207f43277b57cca9673b9ce","_number":762042,"virtual_id_number":762042,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 11:59:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-11-10 10:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-10 11:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-10 11:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 10:12:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-10 11:41:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-10 11:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0a36feaf3cbcbbed704bf90b0a886e9f3136dce","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-11-10 05:11:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2edcf8ee79f58c7264eebc3e5061f78379dfabee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 05:28:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/198a916c801b41248d8d977c78ae9f74 : SUCCESS in 11m 18s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/7fa563290c83442b82cd866c85dbe846 : SUCCESS in 6m 32s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/7cc1c0a177f543a48929e642ca0cefe7 : SUCCESS in 12m 26s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2a73ac34bee4cc5942581f5073d1104 : SUCCESS in 10m 19s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/771ea7850c8e4fcb84fd540f5d7df94d : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1cd68d054194a06bc824227534ed266 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b890686ea28e4361b39fee5e142cb14d : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4a592dc07b45a76eff0ee6aae6e99fc236f8742c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 10:12:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51b4cb2dd482cd14630f8253c5056193ae49024b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-10 11:41:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c080f1b851e19a76989990ada9be393465025587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:41:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57fcfa0e9c42621ea30f41d14965d4a0ac77b9d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:59:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1dbaacae043327d3ccf846ff8a76931c72d9174b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:59:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3991f78dc23f4982b82ce8f1f0d5ae51 : SUCCESS in 9m 34s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/47239750fdb24f1c84f243b6d9052b46 : SUCCESS in 11m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9504d37488304dcaaa5c45c1453a38f1 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ef6d12ea1d4a75838030aaf8e511aa : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd26a44ec4d94dd8b811b8f7c1ed9c40 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"741447bdf598da457207f43277b57cca9673b9ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 12:01:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa2da33824f436d99459acec0ffd215 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e67e37805ede4f1daf59336fd727fee1 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"479d188c29bf9ca6b2b7500c13f5f695f8008ee6","revisions":{"479d188c29bf9ca6b2b7500c13f5f695f8008ee6":{"kind":"REWORK","_number":1,"created":"2020-11-10 05:11:28.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/42/762042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/42/762042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/42/762042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/42/762042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/42/762042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/42/762042/1"}}},"commit":{"parents":[{"commit":"9836375f420e3c6825c4ff8505350a2cf254b42b","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/9836375f420e3c6825c4ff8505350a2cf254b42b"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-11-10 05:11:07.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-11-10 05:11:24.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nChange-Id: Ie74f009f2ef0c56acc49704902095789fb24445a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/479d188c29bf9ca6b2b7500c13f5f695f8008ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/479d188c29bf9ca6b2b7500c13f5f695f8008ee6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
