)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~763996","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~I408b1fa83322fa39a9f9d6dfaed5174ed5a74ae7","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","hashtags":[],"change_id":"I408b1fa83322fa39a9f9d6dfaed5174ed5a74ae7","subject":"Dep\u0027s should be restricted by tox-constraints","status":"ABANDONED","created":"2020-11-24 15:55:41.000000000","updated":"2020-11-27 01:07:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56f20128c08acdef81de48e2357706032e4eadbd","_number":763996,"virtual_id_number":763996,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-24 16:35:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"all":[{"value":-1,"date":"2020-11-26 13:43:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-24 16:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 13:43:51.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"}],"messages":[{"id":"2fffc06eeb26de21bc296d4aebee5ef1c513441c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-24 15:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"046a6422b21247826c1749394466645f15b414a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:35:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9fd6683db82146a0878e2704d491c0cf : SUCCESS in 9m 15s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ae9c5f6b9ba64b3cb316585a3dba56e0 : SUCCESS in 8m 05s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/402cee02932349b99b1cdc788f66c474 : SUCCESS in 9m 41s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/157504ea86c64b8a996ee35d6c65b46d : SUCCESS in 9m 07s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50f889214d2a447b907b4e16ea0193ad : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fd2b92bc39410ab9953d2c2b067e6b : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3c95900d41c4ae28b78fd432a15f5da : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"20e4d04a79158d129bed036247022d5b75e07a88","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-11-26 13:43:51.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"56f20128c08acdef81de48e2357706032e4eadbd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-27 01:07:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"415f4515f6c6efe2ccbb89740d828494ff49b27d","revisions":{"415f4515f6c6efe2ccbb89740d828494ff49b27d":{"kind":"REWORK","_number":1,"created":"2020-11-24 15:55:41.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/96/763996/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/96/763996/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/96/763996/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/96/763996/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/96/763996/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/96/763996/1"}}},"commit":{"parents":[{"commit":"479d188c29bf9ca6b2b7500c13f5f695f8008ee6","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/479d188c29bf9ca6b2b7500c13f5f695f8008ee6"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 14:25:08.000000000","tz":0},"committer":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 15:54:00.000000000","tz":0},"subject":"Dep\u0027s should be restricted by tox-constraints","message":"Dep\u0027s should be restricted by tox-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s tox-constraints.txt\n\nChange-Id: I408b1fa83322fa39a9f9d6dfaed5174ed5a74ae7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/415f4515f6c6efe2ccbb89740d828494ff49b27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/415f4515f6c6efe2ccbb89740d828494ff49b27d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
