)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~768634","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~I71e925f71b7cc3d2104a7b894fe3b3dcfddfaa8d","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","hashtags":[],"change_id":"I71e925f71b7cc3d2104a7b894fe3b3dcfddfaa8d","subject":"Add Victoria release info","status":"ABANDONED","created":"2020-12-29 02:27:24.000000000","updated":"2021-01-28 16:48:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"478d1067d6fb64fb5af5a02d24aabe40591052d6","_number":768634,"virtual_id_number":768634,"owner":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-29 02:48:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 02:48:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46757e45a8600698dfb880c33ac59b0f7773ea20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2020-12-29 02:27:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b54c844f26c003ae21aed08a7521c492cb121d8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2020-12-29 02:36:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2c737e48b3262ed71f8244877399ce85468ced58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 02:48:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5ec8156af0da4ca3b6f6ef8624adb9ef : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e982c86ae84b24b82e8a5196da3507 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee7a4baf727b43cda1f5f3ea8e6ad39e : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"478d1067d6fb64fb5af5a02d24aabe40591052d6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-28 16:48:23.000000000","message":"Abandoned\n\nwe don\u0027t publish reno for these repos","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e993328a4e14435bc540f436f5af3ee2c2f277f1","revisions":{"dca771ecacdad9f8165c509d35af1fc5ea27ba95":{"kind":"REWORK","_number":1,"created":"2020-12-29 02:27:24.000000000","uploader":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"ref":"refs/changes/34/768634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/34/768634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/1"}}},"commit":{"parents":[{"commit":"479d188c29bf9ca6b2b7500c13f5f695f8008ee6","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/479d188c29bf9ca6b2b7500c13f5f695f8008ee6"}]}],"author":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2020-12-29 02:26:52.000000000","tz":480},"committer":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2020-12-29 02:27:06.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nChange-Id: I71e925f71b7cc3d2104a7b894fe3b3dcfddfaa8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/dca771ecacdad9f8165c509d35af1fc5ea27ba95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/dca771ecacdad9f8165c509d35af1fc5ea27ba95"}]},"branch":"refs/heads/master"},"e993328a4e14435bc540f436f5af3ee2c2f277f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-29 02:36:29.000000000","uploader":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"ref":"refs/changes/34/768634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/34/768634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/34/768634/2"}}},"commit":{"parents":[{"commit":"479d188c29bf9ca6b2b7500c13f5f695f8008ee6","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/479d188c29bf9ca6b2b7500c13f5f695f8008ee6"}]}],"author":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2020-12-29 02:26:52.000000000","tz":480},"committer":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2020-12-29 02:35:48.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: I71e925f71b7cc3d2104a7b894fe3b3dcfddfaa8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/e993328a4e14435bc540f436f5af3ee2c2f277f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/e993328a4e14435bc540f436f5af3ee2c2f277f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
