)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~768648","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~stable%2Fvictoria~I5ad63437e200027cfc7e36dce605ef2affc9777b","project":"openstack/openstack-ansible-apt_package_pinning","branch":"stable/victoria","topic":"master","hashtags":[],"change_id":"I5ad63437e200027cfc7e36dce605ef2affc9777b","subject":"Update .gitreview for stable/victoria","status":"ABANDONED","created":"2020-12-29 02:49:05.000000000","updated":"2020-12-29 02:58:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4f8367e367ce1e8d74d5e31f143ff7cd2016487","_number":768648,"virtual_id_number":768648,"owner":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-29 02:57:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 02:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1935e39bf218e4ec0b35baffcd454e30b862bb81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2020-12-29 02:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5910dba4dcd7392c0fa6f2bf36804344492fb34f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 02:57:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/53b22177e2a54dd1a3615003d097ebfb : FAILURE in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f8c53bd2d240dc80db367e4099b003 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f4f8367e367ce1e8d74d5e31f143ff7cd2016487","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"date":"2020-12-29 02:58:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af23aa1b3e90f04da621cb85408c14edf0245fe1","revisions":{"af23aa1b3e90f04da621cb85408c14edf0245fe1":{"kind":"REWORK","_number":1,"created":"2020-12-29 02:49:05.000000000","uploader":{"_account_id":31825,"name":"Xinxin Shen","email":"shenxinxin@inspur.com","username":"XinxinShen"},"ref":"refs/changes/48/768648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/48/768648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/48/768648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/48/768648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/48/768648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/48/768648/1"}}},"commit":{"parents":[{"commit":"e993328a4e14435bc540f436f5af3ee2c2f277f1","subject":"Add Victoria release info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/e993328a4e14435bc540f436f5af3ee2c2f277f1"}]}],"author":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2020-12-29 02:47:47.000000000","tz":480},"committer":{"name":"XinxinShen","email":"shenxinxin@inspur.com","date":"2020-12-29 02:48:05.000000000","tz":480},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nUpdate .gitreview for stable/victoria\n\nChange-Id: I5ad63437e200027cfc7e36dce605ef2affc9777b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/af23aa1b3e90f04da621cb85408c14edf0245fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/af23aa1b3e90f04da621cb85408c14edf0245fe1"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
