)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~868801","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~stable%2Fyoga~I3884174e6bf838959a541851a76964b6173a8966","project":"openstack/openstack-ansible-apt_package_pinning","branch":"stable/yoga","topic":"osa/tox4","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-29 18:17:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3884174e6bf838959a541851a76964b6173a8966","subject":"Update tox.ini to work with 4.0","status":"MERGED","created":"2022-12-28 21:25:35.000000000","updated":"2022-12-29 18:18:05.000000000","submitted":"2022-12-29 18:17:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868801-osa/tox4","meta_rev_id":"b8f7386643b3affdf4794cc48fcebed13537faac","_number":868801,"virtual_id_number":868801,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-29 18:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-12-29 16:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-29 18:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-29 18:02:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 21:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-29 18:02:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"478a80919f9a1028a1a6fb3f9af8f38b70a676b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-28 21:25:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8781403f086b4159f94649c77660b6d943e48dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 21:50:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/314cca12b89c4944a282cbad5209ec52\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/292dd4d051764847a4f8acd52c78cab7 : SUCCESS in 9m 51s\n- openstack-ansible-functional-centos-8-stream https://zuul.opendev.org/t/openstack/build/550cac85bc3e4d519d0fe0d9a9a905c5 : SUCCESS in 7m 24s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0421eebac2aa48e4a47aef63a8c3a8c1 : FAILURE in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff8c3ec41994b5a9ffff56192c3ebdb : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8999a5ef259640babb3776ad8e482600 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"292e28e3c15c543babcb28e209eb341afaa0d76a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-29 13:10:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55fd636434ba89127417e7485da7dc5170af9baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 13:27:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d398a273d6a4bdcb819b5bcc9115b2c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac5e41f2dcd94476b0e6949e098555f3 : SUCCESS in 9m 49s\n- openstack-ansible-functional-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ea1ac4c82fd438db3d16df1e3d1d1d4 : SUCCESS in 7m 00s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7633e77268764fff9912563592285266 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5493d8659dc34867aaa8ee78ceb6bc70 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14b1a88e078b477fb4e6103b5eaf7f40 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"02fa7cda09087631b04fce612d2b007c199eebe6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-29 16:16:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6c2ca3acaf5845f63b0fb5e9cba6d3df30cf433","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-12-29 18:02:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4ef963e4bf1ba6a9e08dd3cb65345637f690930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:04:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"001da8457c74e48e0840e449e3de643a86b66033","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:17:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14cc673cf9dd4b128505475f634d9c27\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06bd525779e74016956ea5e4a5605e7f : SUCCESS in 8m 38s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24c02cdfe25d42e289732daf77157a7a : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca2d61133274d7ebf8c89bb29db424f : SUCCESS in 2m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c2dbc51b3924b5592704a4f01b14652 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dfe340f75ac94316c0b03862cfe08f7f1aaadbca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:17:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b8f7386643b3affdf4794cc48fcebed13537faac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 18:18:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/814cfb62e9fe4ab3b77cf61125c38a58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf81d6b9c9347e58f01b34a25d626a9 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"561fd8c473b7a0af4e9003316ea0e802371065f7","revisions":{"561fd8c473b7a0af4e9003316ea0e802371065f7":{"kind":"REWORK","_number":1,"created":"2022-12-28 21:25:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/01/868801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/01/868801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/01/868801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/01/868801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/01/868801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/01/868801/1"}}},"commit":{"parents":[{"commit":"150a5a0bbce8a6379118f52d0a55aa506db50d86","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/150a5a0bbce8a6379118f52d0a55aa506db50d86"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-28 21:25:33.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-28 21:25:33.000000000","tz":60},"subject":"Update tox.ini to work with 4.0","message":"Update tox.ini to work with 4.0\n\nWith tox release of 4.0, some parameters were deprecated and are ignored now\nwhich causes tox failures. One of the most spread issues we have is using\n`whitelist_externals` isntead of `allowlist_externals`\n\n\nChange-Id: I3884174e6bf838959a541851a76964b6173a8966","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/561fd8c473b7a0af4e9003316ea0e802371065f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/561fd8c473b7a0af4e9003316ea0e802371065f7"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
