)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~948714","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~Ia77586872884a7cbbd2a0c4f645a816969121fd3","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","attention_set":{},"removed_from_attention_set":{"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2025-05-08 14:17:11.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-06 10:44:42.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-05-06 19:33:37.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-05-06 19:09:02.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ia77586872884a7cbbd2a0c4f645a816969121fd3","subject":"Remove tags from README","status":"MERGED","created":"2025-05-02 16:34:37.000000000","updated":"2025-05-08 14:18:25.000000000","submitted":"2025-05-08 14:17:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948714","meta_rev_id":"3bff8793118aa38100ee5840ef00bb79917a8641","_number":948714,"virtual_id_number":948714,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-08 14:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-05-06 10:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-06 19:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-06 19:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-02 16:37:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-05-02 16:37:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-05-02 16:51:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-06 10:47:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"d27c9b8a1838631ac39496f8f8e7680567e56da1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 16:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21dd169b0bc7a9b3f1686e31d0124a973448a4dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 16:37:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8e4d8ed55d03c6f326d874acc9217c6096db1e4e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 16:37:11.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"d53fba7bca6dcb14ec4b89b2274c496bde55275d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 16:40:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fee4b64a530853f25f01386b0c4f8639611baa4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 16:51:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4541ceafd7d34741ad8e7cf61fd94aaf\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dc3cec632cb49bea472109d8fcac0b8 : FAILURE in 5m 26s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b4ffa00054142e4be75bd4bb7a7b190 : SUCCESS in 4m 24s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/40c1befff4dd45aa9bc9332e66edccb9 : SUCCESS in 4m 01s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ef701e544a5450aa51f7911592d6da1 : SUCCESS in 4m 05s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3314c6c14da45a8b2a2cf61db102d9e : SUCCESS in 4m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eadc928fe5849b6b9b6463c956f0a4e : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6312bf1cb8d8441b87087b101f76ab92 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"3f38ac1504b9e69ebb3f72686bd9d19e9d75ba93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-02 17:12:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c2e1ecdaae6026b39c3454ac96f405aecfdc5584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 17:24:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16781e559d3b424d93eeac27648da74a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05fe66af9bd54b59b1a1357c2affdda4 : SUCCESS in 6m 28s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/1baa967d1a4545c59d158c0e9ccd9401 : SUCCESS in 4m 21s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/1e0cda6bb9f64b2b87605f6ddce3fec4 : SUCCESS in 4m 22s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eca84f3bbc884d3098e55eea5a6ef993 : SUCCESS in 4m 01s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f235be7a7934e38a1a8bcbfaab33d22 : SUCCESS in 3m 36s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea1014a6ffc048daa306e542a09f49af : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28205252c9c14bee8e4de047ea5595cd : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":4},{"id":"5e6731cd3c3a0bd6b0811804814d35018dba168c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-06 10:44:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dfb440c5a9e4d51d7c2556bdfe9b9cc5ab890cb8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-06 19:09:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f0a5eebfe5f29336c5dac911289bb89b0db25554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:08:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d80b44cf80a36f2e959959ab52c85de2eb48e514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:17:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b6fac650d14a84a4882c28d129f75f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bba0a347a02d47f8a6c9b3bf8d156dfc : SUCCESS in 7m 04s\n- openstack-ansible-tox-molecule-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6017edb3435476f97faa8e809925737 : SUCCESS in 4m 24s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/a13d81db980f4b02b32ad46ce786d578 : SUCCESS in 3m 59s\n- openstack-ansible-tox-molecule-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3da3dbdf413043f9a6e6c369917b349c : SUCCESS in 5m 18s\n- openstack-ansible-tox-molecule-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcbc0d56b843475ea2ee6afad3115eef : SUCCESS in 4m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba2d7835edc64c2fb00fc531dc9f714f : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c12a340cbab41a3a3f97f56e665e565 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":4},{"id":"6702f5298f6ac9c10da3613eb6288cfe62eab4ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:17:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3bff8793118aa38100ee5840ef00bb79917a8641","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:18:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa60bb5729944999a1d6d6330511116\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae02f77f14b45529e225b6cc3729263 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e4e3b8cf3a005629d09150f20486f9c946271f77","revisions":{"80d6dc7c446832ddb42298e9b9878082a4c1024b":{"kind":"REWORK","_number":1,"created":"2025-05-02 16:34:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/948714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/14/948714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/1"}}},"commit":{"parents":[{"commit":"f99bc130d4758d65eb95bac883770d9bd41f7d2e","subject":"Use OSA_TEST_REQUIREMENTS_FILE for molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/f99bc130d4758d65eb95bac883770d9bd41f7d2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:34:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:34:37.000000000","tz":0},"subject":"Remove tags from README","message":"Remove tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: Ia77586872884a7cbbd2a0c4f645a816969121fd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/80d6dc7c446832ddb42298e9b9878082a4c1024b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/80d6dc7c446832ddb42298e9b9878082a4c1024b"}]},"branch":"refs/heads/master"},"ca5f1d8ba211b9f8085e863cbb73c88382a98ccc":{"kind":"REWORK","_number":2,"created":"2025-05-02 16:37:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/948714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/14/948714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/2"}}},"commit":{"parents":[{"commit":"f99bc130d4758d65eb95bac883770d9bd41f7d2e","subject":"Use OSA_TEST_REQUIREMENTS_FILE for molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/f99bc130d4758d65eb95bac883770d9bd41f7d2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:34:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:37:03.000000000","tz":0},"subject":"Remove tags from README","message":"Remove tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: Ia77586872884a7cbbd2a0c4f645a816969121fd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/ca5f1d8ba211b9f8085e863cbb73c88382a98ccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/ca5f1d8ba211b9f8085e863cbb73c88382a98ccc"}]},"branch":"refs/heads/master"},"9018e38c39d6e48f620a511641583aa7b525100c":{"kind":"REWORK","_number":3,"created":"2025-05-02 16:40:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/948714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/14/948714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/3"}}},"commit":{"parents":[{"commit":"f99bc130d4758d65eb95bac883770d9bd41f7d2e","subject":"Use OSA_TEST_REQUIREMENTS_FILE for molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/f99bc130d4758d65eb95bac883770d9bd41f7d2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:34:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:40:12.000000000","tz":0},"subject":"Remove tags from README","message":"Remove tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: Ia77586872884a7cbbd2a0c4f645a816969121fd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/9018e38c39d6e48f620a511641583aa7b525100c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/9018e38c39d6e48f620a511641583aa7b525100c"}]},"branch":"refs/heads/master"},"e4e3b8cf3a005629d09150f20486f9c946271f77":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-05-02 17:12:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/948714/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/14/948714/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/14/948714/4"}}},"commit":{"parents":[{"commit":"ca273b552bed36d814c431edaf9e7e0528845d27","subject":"Use FQCN for modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/ca273b552bed36d814c431edaf9e7e0528845d27"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 16:34:37.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-02 17:12:08.000000000","tz":0},"subject":"Remove tags from README","message":"Remove tags from README\n\nThe tags framework has been discontinued for a long time.\n\nhttps://governance.openstack.org/tc/reference/tags/\n\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nChange-Id: Ia77586872884a7cbbd2a0c4f645a816969121fd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/e4e3b8cf3a005629d09150f20486f9c946271f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/e4e3b8cf3a005629d09150f20486f9c946271f77"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
