)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~985561","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~master~Ic56fc64fcdd2241025b2d016ad802ad7474d2976","project":"openstack/openstack-ansible-apt_package_pinning","branch":"master","topic":"openstack/openstack-ansible-sync/sync-files","hashtags":[],"change_id":"Ic56fc64fcdd2241025b2d016ad802ad7474d2976","subject":"Updated from OpenStack Ansible Integrated repository","status":"MERGED","created":"2026-04-20 23:14:07.000000000","updated":"2026-04-21 08:51:59.000000000","submitted":"2026-04-21 08:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985561-openstack/openstack-ansible-sync/sync-files","meta_rev_id":"ba08b86692a5e5dddd394c1f709603f68b444df4","_number":985561,"virtual_id_number":985561,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 08:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-04-21 08:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-21 08:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 23:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 08:12:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"da48d9ff3656418c5b6fa4e456c70f0bd7f9dcdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-04-20 23:14:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54fa79486296f924b8b9b0f14518ac4b3ee35b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:40:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec1caea1c0b248c3bd724058e00c008f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32c23813c90d4ac5a8b7edc908926907 : SUCCESS in 10m 57s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6daccbc223d44af58b10cfaeb0664132 : SUCCESS in 3m 44s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/36d895f0fdd542608f641fb7f37f2f5f : SUCCESS in 3m 30s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b086617b95f40d68086a0f25a7386f6 : SUCCESS in 5m 33s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/29512576180447898c6e24f90e6b06f8 : SUCCESS in 4m 57s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d4e4d6c044e4b73b233b6338d812655 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f437a775f66f48ff9065005d425433ad : SUCCESS in 2m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/957db949415b4f86a6468dd5df28e4ba : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7c562ce3a0028a2b76bfa5c5348272a591aef31c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 08:12:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f0099a1864d5ed6cf29d76a865bd8f45ab26f40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:15:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3777da7618888c2af03ad388cf1c49847056b0e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:49:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ded1fa91178545fa8ffedbcb2946394a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac88b7ac95b6475b8fca5787a84861bb : SUCCESS in 16m 36s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/2ab4442cf386495a8a6bd2a0a7b30d8a : SUCCESS in 5m 50s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/efddb4fa43064307a9436f3c5c709fb3 : SUCCESS in 5m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1694cf5eb32d4b0fba2fd40d72eed2a5 : SUCCESS in 5m 31s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1f35d71430f144eb8009b7937a80b4c9 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d300acdf98664ba5a7615ad01068e75b : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ac15392e8954f4e874a8cb3c701f3fd : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a73ea473864e66cef03cba5e1b9cf3c74de77942","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:49:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba08b86692a5e5dddd394c1f709603f68b444df4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd382cd31bbd4d6f9e386cc81cfb4875\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0cb963f96845398f045e72f2589106 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74fbcde4e444423ed7af017e3127c5f10f7f8ee","revisions":{"f74fbcde4e444423ed7af017e3127c5f10f7f8ee":{"kind":"REWORK","_number":1,"created":"2026-04-20 23:14:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/985561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/61/985561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/61/985561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/61/985561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/61/985561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/61/985561/1"}}},"commit":{"parents":[{"commit":"5049bfb8d7ca463475c9fb37a15add67f07acecd","subject":"Merge \"Add CONTRIBUTING.rst file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/5049bfb8d7ca463475c9fb37a15add67f07acecd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:14:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-04-20 23:14:06.000000000","tz":0},"subject":"Updated from OpenStack Ansible Integrated repository","message":"Updated from OpenStack Ansible Integrated repository\n\nChange-Id: Ic56fc64fcdd2241025b2d016ad802ad7474d2976\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:playbooks/proposal/propose_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/f74fbcde4e444423ed7af017e3127c5f10f7f8ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/f74fbcde4e444423ed7af017e3127c5f10f7f8ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
