)]}'
{"id":"openstack%2Fopenstack-ansible-apt_package_pinning~992070","triplet_id":"openstack%2Fopenstack-ansible-apt_package_pinning~stable%2F2026.1~I9c002b0d7130f4ebf4c47c84c1591e48a6d2c94f","project":"openstack/openstack-ansible-apt_package_pinning","branch":"stable/2026.1","topic":"create-2026.1","hashtags":[],"change_id":"I9c002b0d7130f4ebf4c47c84c1591e48a6d2c94f","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","status":"MERGED","created":"2026-06-08 09:06:55.000000000","updated":"2026-06-08 20:16:49.000000000","submitted":"2026-06-08 20:14:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992070-create-2026.1","meta_rev_id":"bfa2aa9b32e1bbad44cd2651ad8f2dd07fe4db61","_number":992070,"virtual_id_number":992070,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-08 20:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-08 20:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-08 20:04:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 09:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-08 20:04:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"099b1411ee4058cb0da4af928fb3cea1a2d79221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-06-08 09:06:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc32de1e32bf2195d57e388ed0ab4f23a2a0a6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 09:29:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b3972e4ff60432bbc47891119d9fa0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a264962af04b42bc4bc5fe74bfa025 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be6a82bc43fb42abb691b0ef9477c52c : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6fc6eac3ae896f4b69f654646c9a4c25dff4a5c8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-08 20:04:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfaa3af560731349e446ee18a52a8d3acaa3735a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 20:05:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2865142ca841b60d268ba1906c76dac7e959996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 20:14:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6740d0d35b0b4f2aa075dae0a093696c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb0eb44d7ab4d16a38195c2df5f070c : SUCCESS in 2m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8456bcb10903439397067ad648aba08e : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"027cd7202eef4259816792e0a085ac11cc58dda9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 20:14:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bfa2aa9b32e1bbad44cd2651ad8f2dd07fe4db61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 20:16:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/081141e37f59452abaf648d46c11cd73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d14b30d3a484d49a53b0f21140022b1 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e296a75100fcfa17953b76b2a138c075cdecc6f","revisions":{"9e296a75100fcfa17953b76b2a138c075cdecc6f":{"kind":"REWORK","_number":1,"created":"2026-06-08 09:06:55.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/70/992070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning","ref":"refs/changes/70/992070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/70/992070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/70/992070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/70/992070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-apt_package_pinning refs/changes/70/992070/1"}}},"commit":{"parents":[{"commit":"fa1099225b7b9e24ac3609bdf345c8c3671d2612","subject":"Update .gitreview for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/fa1099225b7b9e24ac3609bdf345c8c3671d2612"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-06-08 09:06:54.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-06-08 09:06:54.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I9c002b0d7130f4ebf4c47c84c1591e48a6d2c94f\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/9e296a75100fcfa17953b76b2a138c075cdecc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-apt_package_pinning/commit/9e296a75100fcfa17953b76b2a138c075cdecc6f"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
