)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~287534","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I5f28e0cafc358b868d633acc0628a93b5cd0ab48","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"vagrant_local_testing","hashtags":[],"change_id":"I5f28e0cafc358b868d633acc0628a93b5cd0ab48","subject":"Adding Vagrantfile for local developer testing","status":"MERGED","created":"2016-03-03 00:57:42.000000000","updated":"2016-03-18 08:52:22.000000000","submitted":"2016-03-18 08:52:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc7617de0dd4deb34f5b8d4925c5ccf633ce5393","_number":287534,"virtual_id_number":287534,"owner":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2016-03-18 08:52:22.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2016-03-11 07:46:50.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-03-11 17:46:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2016-03-18 08:50:27.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-11 17:46:22.000000000","updated_by":{"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"},"reviewer":{"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"},"state":"REVIEWER"},{"updated":"2016-03-18 08:50:27.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2016-03-18 08:52:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"6329d4a6e8de50d592dd51772fe549973ade1590","author":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"date":"2016-03-03 00:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85a3010d30e8ccc0ba2cd17f5916a7946d51f789","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-03 01:00:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-galera_client-docs http://docs-draft.openstack.org/34/287534/1/check/gate-openstack-ansible-galera_client-docs/c783470//doc/build/html/ : SUCCESS in 39s\n- gate-openstack-ansible-galera_client-linters http://logs.openstack.org/34/287534/1/check/gate-openstack-ansible-galera_client-linters/509714d/ : SUCCESS in 41s\n- gate-openstack-ansible-galera_client-dsvm-ansible-functional-trusty http://logs.openstack.org/34/287534/1/check/gate-openstack-ansible-galera_client-dsvm-ansible-functional-trusty/472749b/ : SUCCESS in 1m 26s\n- gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv http://logs.openstack.org/34/287534/1/check/gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv/9dbf77c/ : FAILURE in 1m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f466944dc8c12e3f827c9c6bf40cfb74a1fc4b37","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2016-03-03 11:25:38.000000000","message":"Patch Set 1: Code-Review+2\n\nTested it - works for me!","accounts_in_message":[],"_revision_number":1},{"id":"6ed0b9eefd2377c888535e00da88d629ac1d671a","author":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"date":"2016-03-08 17:51:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"64f236904d33e599b0347479717bb4c849c021fe","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-08 17:57:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-openstack-ansible-galera_client-docs http://docs-draft.openstack.org/34/287534/2/check/gate-openstack-ansible-galera_client-docs/f23a566//doc/build/html/ : SUCCESS in 3m 04s\n- gate-openstack-ansible-galera_client-linters http://logs.openstack.org/34/287534/2/check/gate-openstack-ansible-galera_client-linters/9f7ebb9/ : SUCCESS in 32s\n- gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty http://logs.openstack.org/34/287534/2/check/gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty/d578f20/ : FAILURE in 2m 44s\n- gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv http://logs.openstack.org/34/287534/2/check/gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv/6489593/ : FAILURE in 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"350d546c9888d3c7b0125889cddf342b84e62594","author":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"date":"2016-03-08 19:49:15.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"79b6ec054135a933761ebdc2b806e1af7dddef89","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-08 19:52:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-galera_client-docs http://docs-draft.openstack.org/34/287534/3/check/gate-openstack-ansible-galera_client-docs/302f228//doc/build/html/ : SUCCESS in 1m 05s\n- gate-openstack-ansible-galera_client-linters http://logs.openstack.org/34/287534/3/check/gate-openstack-ansible-galera_client-linters/f7c5af6/ : SUCCESS in 31s\n- gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty http://logs.openstack.org/34/287534/3/check/gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty/d46ea51/ : SUCCESS in 1m 06s\n- gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv http://logs.openstack.org/34/287534/3/check/gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv/e14f0cf/ : FAILURE in 1m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3bd721ed45613e1aced9cf4968ab008de50a0d45","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-11 07:00:44.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"cecde5cb3800603cd3f20077ff6026d9bb808aa2","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2016-03-11 07:46:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8aabb0ca122523b0c1e5c8b3e031b6357a1a4363","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-11 07:49:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-galera_client-docs http://docs-draft.openstack.org/34/287534/4/check/gate-openstack-ansible-galera_client-docs/43d9aa3//doc/build/html/ : SUCCESS in 45s\n- gate-openstack-ansible-galera_client-linters http://logs.openstack.org/34/287534/4/check/gate-openstack-ansible-galera_client-linters/4ef46be/ : SUCCESS in 43s\n- gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty http://logs.openstack.org/34/287534/4/check/gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty/7b24ebd/ : SUCCESS in 1m 21s\n- gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv http://logs.openstack.org/34/287534/4/check/gate-openstack-ansible-galera_client-dsvm-ansible-functional-centos7-nv/363c294/ : FAILURE in 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c80208da3c17bc9ac481f89d11afc36e396031c","author":{"_account_id":12807,"name":"Steve Lewis (stevelle)","email":"stevelle@gmail.com","username":"stevelle"},"date":"2016-03-11 17:46:22.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nHaven\u0027t spun this yet.\nOne nit noted.","accounts_in_message":[],"_revision_number":4},{"id":"1c8a693baab9e829cf89d0bddaa4584c2b2650d6","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2016-03-18 08:50:27.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1d62a5b31e9e0791a13bf8d1481127599c3719d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-18 08:50:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"472b998786bcd32ab9bde42749aadc1ff2f49248","date":"2016-03-18 08:52:22.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4},{"id":"f04a5f29309628c3f8659112a73015948b8f51dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-18 08:52:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-galera_client-docs http://docs-draft.openstack.org/34/287534/4/gate/gate-openstack-ansible-galera_client-docs/db0baa0//doc/build/html/ : SUCCESS in 48s\n- gate-openstack-ansible-galera_client-linters http://logs.openstack.org/34/287534/4/gate/gate-openstack-ansible-galera_client-linters/5ac5c8f/ : SUCCESS in 45s\n- gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty http://logs.openstack.org/34/287534/4/gate/gate-openstack-ansible-galera_client-ansible-func-ubuntu-trusty/c74ca64/ : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"518e63de266856af97de7047156b68644e19cf02","revisions":{"8d1fa1f0f05a9191f7e6e4a4b4dd1cc4004ef0d2":{"kind":"REWORK","_number":1,"created":"2016-03-03 00:57:42.000000000","uploader":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"ref":"refs/changes/34/287534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/34/287534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/1"}}},"commit":{"parents":[{"commit":"58a35de92a0567a1b72c8ce588d1e8bf5632e249","subject":"Only update apt cache if necessary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/58a35de92a0567a1b72c8ce588d1e8bf5632e249"}]}],"author":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-03 00:57:11.000000000","tz":-300},"committer":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-03 00:57:11.000000000","tz":-300},"subject":"Adding Vagrantfile for local developer testing","message":"Adding Vagrantfile for local developer testing\n\nThis Vagrantfile attempts to replicate the gate check\ntesting to the extent possible to allow developers to\nrun all tests locally in a Vagrant environment prior\nto submitting changes for review.\n\nContributor documentation updated to demonstrate usage\nof the Vagrantfile.\n\nChange-Id: I5f28e0cafc358b868d633acc0628a93b5cd0ab48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/8d1fa1f0f05a9191f7e6e4a4b4dd1cc4004ef0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/8d1fa1f0f05a9191f7e6e4a4b4dd1cc4004ef0d2"}]},"branch":"refs/heads/master"},"7bdd0adcab0ca1378db2462158d16bace0fd041c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2016-03-08 17:51:47.000000000","uploader":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"ref":"refs/changes/34/287534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/34/287534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/2"}}},"commit":{"parents":[{"commit":"2ddb24ebc37988ead8e42943c174704192fa82e2","subject":"Removing unneeded with_items usage for clarity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/2ddb24ebc37988ead8e42943c174704192fa82e2"}]}],"author":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-03 00:57:11.000000000","tz":-300},"committer":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-08 17:51:24.000000000","tz":-300},"subject":"Adding Vagrantfile for local developer testing","message":"Adding Vagrantfile for local developer testing\n\nThis Vagrantfile attempts to replicate the gate check\ntesting to the extent possible to allow developers to\nrun all tests locally in a Vagrant environment prior\nto submitting changes for review.\n\nContributor documentation updated to demonstrate usage\nof the Vagrantfile.\n\nChange-Id: I5f28e0cafc358b868d633acc0628a93b5cd0ab48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/7bdd0adcab0ca1378db2462158d16bace0fd041c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/7bdd0adcab0ca1378db2462158d16bace0fd041c"}]},"branch":"refs/heads/master"},"2de92c35af2f8cef390a1a2b81cf67b9bc272852":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-03-08 19:49:15.000000000","uploader":{"_account_id":19814,"name":"Travis Truman (automagically)","email":"travis_truman@cable.comcast.com","username":"trumant"},"ref":"refs/changes/34/287534/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/34/287534/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/3"}}},"commit":{"parents":[{"commit":"d5502a9dcce1e07fdc3762ef9683aa733645132d","subject":"Initial commit for bindep requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/d5502a9dcce1e07fdc3762ef9683aa733645132d"}]}],"author":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-03 00:57:11.000000000","tz":-300},"committer":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-08 19:49:06.000000000","tz":-300},"subject":"Adding Vagrantfile for local developer testing","message":"Adding Vagrantfile for local developer testing\n\nThis Vagrantfile attempts to replicate the gate check\ntesting to the extent possible to allow developers to\nrun all tests locally in a Vagrant environment prior\nto submitting changes for review.\n\nContributor documentation updated to demonstrate usage\nof the Vagrantfile.\n\nChange-Id: I5f28e0cafc358b868d633acc0628a93b5cd0ab48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/2de92c35af2f8cef390a1a2b81cf67b9bc272852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/2de92c35af2f8cef390a1a2b81cf67b9bc272852"}]},"branch":"refs/heads/master"},"518e63de266856af97de7047156b68644e19cf02":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2016-03-11 07:46:50.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/34/287534/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/34/287534/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/34/287534/4"}}},"commit":{"parents":[{"commit":"b5b767a5aa876869422e4c698c70451932dac1a4","subject":"Ensure apt-transport-https is present before installing from secure repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/b5b767a5aa876869422e4c698c70451932dac1a4"}]}],"author":{"name":"Travis Truman","email":"travis_truman@cable.comcast.com","date":"2016-03-03 00:57:11.000000000","tz":-300},"committer":{"name":"Jesse Pretorius (odyssey4me)","email":"jesse.pretorius@rackspace.co.uk","date":"2016-03-11 07:46:50.000000000","tz":0},"subject":"Adding Vagrantfile for local developer testing","message":"Adding Vagrantfile for local developer testing\n\nThis Vagrantfile attempts to replicate the gate check\ntesting to the extent possible to allow developers to\nrun all tests locally in a Vagrant environment prior\nto submitting changes for review.\n\nContributor documentation updated to demonstrate usage\nof the Vagrantfile.\n\nChange-Id: I5f28e0cafc358b868d633acc0628a93b5cd0ab48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/518e63de266856af97de7047156b68644e19cf02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/518e63de266856af97de7047156b68644e19cf02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
