)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~654337","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~Ib7edcdfaf2d68b744c232c00c7d38df6f4c437de","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ib7edcdfaf2d68b744c232c00c7d38df6f4c437de","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:12:18.000000000","updated":"2019-04-22 11:34:56.000000000","submitted":"2019-04-22 11:34:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654337-1555932896929-11ac1cc1","meta_rev_id":"313061bcd4823bee6b9694e2ec7bb2681933cc61","_number":654337,"virtual_id_number":654337,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 11:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 11:04:42.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 11:04:42.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 11:04:42.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 11:34:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85400645963a6778302d113618938cc40b8c03e0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:12:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1f4980981320d8b3f844155850a8c8585194b95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:38:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/654337/1/check/openstack-ansible-linters/acca383/ : SUCCESS in 5m 02s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/37/654337/1/check/openstack-ansible-functional-debian-stable/f281ed5/ : SUCCESS in 6m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/37/654337/1/check/openstack-ansible-functional-centos-7/df716c1/ : SUCCESS in 7m 18s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/37/654337/1/check/openstack-ansible-functional-ubuntu-bionic/6f94c27/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/37/654337/1/check/openstack-tox-docs/080a207/html/ : SUCCESS in 2m 28s\n- openstack-ansible-ssl-galera_server-opensuse-423 http://logs.openstack.org/37/654337/1/check/openstack-ansible-ssl-galera_server-opensuse-423/f523e0a/ : FAILURE in 6m 20s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/37/654337/1/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/0b58c04/ : SUCCESS in 16m 50s (non-voting)\n- openstack-ansible-functional-galera_server-opensuse-423 http://logs.openstack.org/37/654337/1/check/openstack-ansible-functional-galera_server-opensuse-423/646268b/ : FAILURE in 5m 41s (non-voting)\n- openstack-ansible-functional-galera_server-ubuntu-xenial http://logs.openstack.org/37/654337/1/check/openstack-ansible-functional-galera_server-ubuntu-xenial/fb55519/ : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abff3fff5f8538b4b82669120e136844b5f6f9a0","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 11:04:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b3701d0653dc9bfd79f3a3d631ba24fb90b007b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:04:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f820173497637adf1e78ccf3aeb9fb25984cae36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:34:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f03d2f873a25411c09cf68a72cbc8225ade71764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:34:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/37/654337/1/gate/openstack-ansible-linters/6113847/ : SUCCESS in 5m 43s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/37/654337/1/gate/openstack-ansible-functional-debian-stable/781ed34/ : SUCCESS in 5m 44s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/37/654337/1/gate/openstack-ansible-functional-centos-7/1d3a276/ : SUCCESS in 7m 25s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/37/654337/1/gate/openstack-ansible-functional-ubuntu-bionic/bd0d367/ : SUCCESS in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/37/654337/1/gate/openstack-tox-docs/9c1bacd/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d3f35508fab82de4e81f248d5ef84930044846a","revisions":{"6d3f35508fab82de4e81f248d5ef84930044846a":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:12:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/654337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/37/654337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/37/654337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/37/654337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/37/654337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/37/654337/1"}}},"commit":{"parents":[{"commit":"1b2c07e5f242f569123587244c735755bed85c8e","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/1b2c07e5f242f569123587244c735755bed85c8e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:12:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:12:16.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ib7edcdfaf2d68b744c232c00c7d38df6f4c437de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/6d3f35508fab82de4e81f248d5ef84930044846a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/6d3f35508fab82de4e81f248d5ef84930044846a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
