)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~654491","triplet_id":"openstack%2Fopenstack-ansible-galera_client~stable%2Fstein~I0c399cf9dd03193df97e704557e4a9b31c1ab492","project":"openstack/openstack-ansible-galera_client","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I0c399cf9dd03193df97e704557e4a9b31c1ab492","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:36:28.000000000","updated":"2019-04-22 16:28:33.000000000","submitted":"2019-04-22 16:28:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654491-1555950513396-7a5aba1f","meta_rev_id":"85e217c06e095f3d5c35390c5e7a9b7309e650f6","_number":654491,"virtual_id_number":654491,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 16:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:07:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 16:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab3ad78f19e12d9bbb9a5370172c7ddbe2c21e7c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e3a1b01c3d4ca62709d645a42a66c90148a073e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:07:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25b6caf06868a7597246e83edc1b639f9a5486af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/654491/1/check/openstack-ansible-linters/172630b/ : SUCCESS in 6m 10s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/91/654491/1/check/openstack-ansible-functional-debian-stable/79991f9/ : SUCCESS in 6m 18s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/91/654491/1/check/openstack-ansible-functional-centos-7/be09cd6/ : SUCCESS in 7m 38s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/91/654491/1/check/openstack-ansible-functional-ubuntu-bionic/f3015d6/ : SUCCESS in 5m 13s\n- openstack-tox-docs http://logs.openstack.org/91/654491/1/check/openstack-tox-docs/4d8ed34/html/ : SUCCESS in 2m 41s\n- openstack-ansible-ssl-galera_server-opensuse-423 http://logs.openstack.org/91/654491/1/check/openstack-ansible-ssl-galera_server-opensuse-423/27923b9/ : FAILURE in 5m 33s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/91/654491/1/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/083fea2/ : SUCCESS in 15m 59s (non-voting)\n- openstack-ansible-functional-galera_server-opensuse-423 http://logs.openstack.org/91/654491/1/check/openstack-ansible-functional-galera_server-opensuse-423/e7af7cf/ : FAILURE in 10m 26s (non-voting)\n- openstack-ansible-functional-galera_server-ubuntu-xenial http://logs.openstack.org/91/654491/1/check/openstack-ansible-functional-galera_server-ubuntu-xenial/09beb62/ : SUCCESS in 27m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3b7844e09834233be52b08de72b46d2c0b78cf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:10:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cec27fce10d1ef844cac307cb238a7f806f9f6da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:28:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/91/654491/1/gate/openstack-ansible-linters/2a0ddd8/ : SUCCESS in 4m 50s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/91/654491/1/gate/openstack-ansible-functional-debian-stable/28778bd/ : SUCCESS in 6m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/91/654491/1/gate/openstack-ansible-functional-centos-7/ed236e5/ : SUCCESS in 7m 56s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/91/654491/1/gate/openstack-ansible-functional-ubuntu-bionic/cf53ecb/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/91/654491/1/gate/openstack-tox-docs/d1fa719/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e07e11a5e25bec1bfadf2556c9012b6a26bbbac4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:28:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbef8a896550121195e17fb667e01641d2ac8de4","revisions":{"bbef8a896550121195e17fb667e01641d2ac8de4":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:36:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/654491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/91/654491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/91/654491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/91/654491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/91/654491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/91/654491/1"}}},"commit":{"parents":[{"commit":"e312ba116c5f27ed000e9094f87b9ca5f265f975","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/e312ba116c5f27ed000e9094f87b9ca5f265f975"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:36:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:36:26.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I0c399cf9dd03193df97e704557e4a9b31c1ab492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/bbef8a896550121195e17fb667e01641d2ac8de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/bbef8a896550121195e17fb667e01641d2ac8de4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
