)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~657299","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~Id5565bcf0369fc4dca6e48cb2f85ba568ef84795","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"osa/debian-support","hashtags":[],"change_id":"Id5565bcf0369fc4dca6e48cb2f85ba568ef84795","subject":"debian: use ansible_lsb.codename","status":"ABANDONED","created":"2019-05-06 02:28:56.000000000","updated":"2019-12-26 15:36:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce7bf2574019b2bfd455a5f57ee73a2561bbd3ed","_number":657299,"virtual_id_number":657299,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-06 02:49:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 02:49:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18819f32a10f741d2c13f7e56aecf1f48f3eb1f7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-06 02:28:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6673b65afc2886b0c14ed437319d9ca448dcc6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 02:49:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/657299/1/check/openstack-ansible-linters/37e8e0b/ : SUCCESS in 4m 11s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/99/657299/1/check/openstack-ansible-functional-debian-stable/fb42a9d/ : SUCCESS in 6m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/657299/1/check/openstack-ansible-functional-centos-7/4d1c6a5/ : SUCCESS in 8m 00s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/657299/1/check/openstack-ansible-functional-ubuntu-bionic/2cef3b5/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/99/657299/1/check/openstack-tox-docs/16916c2/html/ : SUCCESS in 2m 22s\n- openstack-ansible-ssl-galera_server-opensuse-423 http://logs.openstack.org/99/657299/1/check/openstack-ansible-ssl-galera_server-opensuse-423/5460667/ : FAILURE in 6m 41s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/99/657299/1/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/0d87c14/ : SUCCESS in 17m 10s (non-voting)\n- openstack-ansible-functional-galera_server-opensuse-423 http://logs.openstack.org/99/657299/1/check/openstack-ansible-functional-galera_server-opensuse-423/65711cf/ : FAILURE in 7m 03s (non-voting)\n- openstack-ansible-functional-galera_server-ubuntu-xenial http://logs.openstack.org/99/657299/1/check/openstack-ansible-functional-galera_server-ubuntu-xenial/913dd28/ : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa33de7466b81945e840bad173c112da12bedc1a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-12-26 15:36:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f50cd2f151068c49c659a4104aba41dc4ecbee08","revisions":{"f50cd2f151068c49c659a4104aba41dc4ecbee08":{"kind":"REWORK","_number":1,"created":"2019-05-06 02:28:56.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/99/657299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/99/657299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/99/657299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/99/657299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/99/657299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/99/657299/1"}}},"commit":{"parents":[{"commit":"60f9d61ca8730d4abc500869372226f4940aa0aa","subject":"Merge \"Trivial: Fix the pep8 warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/60f9d61ca8730d4abc500869372226f4940aa0aa"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-06 02:28:29.000000000","tz":-360},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-06 02:28:29.000000000","tz":-360},"subject":"debian: use ansible_lsb.codename","message":"debian: use ansible_lsb.codename\n\nThe LSB codename is far more reliable in order of giving us the\nname of the current series as ansible_distribution_version can\nbe inaccurate in the case of unreleased versions such as Buster\nwhich would return \"NA\" for that variable.\n\nChange-Id: Id5565bcf0369fc4dca6e48cb2f85ba568ef84795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f50cd2f151068c49c659a4104aba41dc4ecbee08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f50cd2f151068c49c659a4104aba41dc4ecbee08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
