)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~660319","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~Ia2d252cfe4a7ef6d10369635a320c208262692d6","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"Ia2d252cfe4a7ef6d10369635a320c208262692d6","subject":"Make coverage consistent","status":"MERGED","created":"2019-05-21 09:42:32.000000000","updated":"2019-05-22 13:30:31.000000000","submitted":"2019-05-22 13:30:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660319-1558531831832-74850baa","meta_rev_id":"b27c4832311101373122c57e47ea3820987140da","_number":660319,"virtual_id_number":660319,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-22 13:30:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-22 12:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-22 10:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-22 12:06:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 10:39:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-22 12:06:40.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-22 13:30:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9abb7262103cd37323cec967a333000ab175f6ee","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 09:42:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c049c9145830a9976809cdafaa3c75cb1b50726","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:42:46.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-ansible-galera_client on branch master.  The\nerror was:\n\n  Job openstack-ansible-ssl-galera_server-opensuse-150 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-ansible-deploy-aio_metal-jobs\n        - openstack-ansible-deploy-aio_distro_metal-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-ansible-ssl-galera_server-opensuse-150\n          - openstack-ansible-ssl-galera_server-ubuntu-xenial\n\n  in \"openstack/openstack-ansible-galera_client/zuul.d/project.yaml@master\", line 16, column 3","accounts_in_message":[],"_revision_number":1},{"id":"59d2174b94d138e52385436e4690a31fc6033745","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 09:56:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"da1168a95f513d07916f349f1a7b992bba45023b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:14:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/660319/2/check/openstack-ansible-deploy-aio_metal-debian-stable/a18e83a/ : SUCCESS in 1h 00m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/660319/2/check/openstack-ansible-deploy-aio_metal-centos-7/767f51a/ : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/19/660319/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/7410173/ : SUCCESS in 1h 00m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/660319/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8f7b40d/ : SUCCESS in 56m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/660319/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3145d7d/ : SUCCESS in 52m 51s\n- openstack-tox-docs http://logs.openstack.org/19/660319/2/check/openstack-tox-docs/a66fc40/html/ : SUCCESS in 3m 00s\n- openstack-ansible-ssl-galera_server-opensuse-15 http://logs.openstack.org/19/660319/2/check/openstack-ansible-ssl-galera_server-opensuse-15/c7a91e5/ : FAILURE in 31m 27s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/19/660319/2/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/766d1ca/ : SUCCESS in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"98d4a37892438726bde6e35b6f8cddf31dbaddfe","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 11:35:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"29565a15a5fc018874589869b92fe52c63622c2a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 12:35:21.000000000","message":"Patch Set 2:\n\nThere seem to be something wrong with the galera_server. It is not related to this aptch, so I think we can solve it later.","accounts_in_message":[],"_revision_number":2},{"id":"f5db1dc03adc6c481bdc36a63f5c727bd40749a4","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 15:14:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d1686c50606a8af56159313b83cb5a180e96e7d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 15:14:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d076ab93fd075218dea523dc72b386f60788106","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 15:17:07.000000000","message":"Patch Set 2: Code-Review-1 -Workflow\n\nNo linters job present","accounts_in_message":[],"_revision_number":2},{"id":"43dd7f21796cdfc53c8047f8ad4c1783e8e72352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:33:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/660319/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/ce20864/ : SUCCESS in 57m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/660319/2/gate/openstack-ansible-deploy-aio_metal-centos-7/0221724/ : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/19/660319/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/0661041/ : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/660319/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/def52ff/ : SUCCESS in 58m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/660319/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/d31a4dd/ : SUCCESS in 50m 48s\n- openstack-tox-docs http://logs.openstack.org/19/660319/2/gate/openstack-tox-docs/ccd4a4a/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d1d4f98b0cfd67c28f7bf36f06eadac113a73481","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 18:33:11.000000000","message":"Removed Code-Review+2 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"8547259b3792274018292fca9d673cb528ee7ba5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 07:20:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e4fc4ccaaab20a20932d2e7a560a313f3602f63b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 07:20:48.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0cab445f9ce7c8c4281bce7e528afd0f0a4adb9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 08:42:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/660319/4/check/openstack-ansible-deploy-aio_metal-debian-stable/9c4b6bc/ : SUCCESS in 58m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/660319/4/check/openstack-ansible-deploy-aio_metal-centos-7/c1ab1c9/ : SUCCESS in 1h 08m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/19/660319/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/0933d3c/ : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/660319/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/82c897d/ : SUCCESS in 55m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/660319/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ad95b8b/ : SUCCESS in 53m 18s\n- openstack-ansible-linters http://logs.openstack.org/19/660319/4/check/openstack-ansible-linters/808c0a6/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/19/660319/4/check/openstack-tox-docs/238cc4e/html/ : SUCCESS in 2m 23s\n- openstack-ansible-ssl-galera_server-opensuse-15 http://logs.openstack.org/19/660319/4/check/openstack-ansible-ssl-galera_server-opensuse-15/e13420d/ : SUCCESS in 35m 31s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/19/660319/4/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/85a076e/ : SUCCESS in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72a0f3af8d9030fd04834cb1975751aa9c846a8d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-22 10:39:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8f1070c575540976525ae99ee3d1f4a675a83eb9","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 12:06:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"42e4939c2e7fa622971be5dbdb8e6028d346c23b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 12:06:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e867343eae64f6d32ed2402ffd399f076330a911","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 13:30:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/660319/4/gate/openstack-ansible-deploy-aio_metal-debian-stable/3cbc1e4/ : SUCCESS in 58m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/660319/4/gate/openstack-ansible-deploy-aio_metal-centos-7/918ea76/ : SUCCESS in 1h 16m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/19/660319/4/gate/openstack-ansible-deploy-aio_metal-opensuse-150/8c119de/ : SUCCESS in 1h 09m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/660319/4/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d3be598/ : SUCCESS in 1h 03m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/660319/4/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5b311a5/ : SUCCESS in 53m 06s\n- openstack-ansible-linters http://logs.openstack.org/19/660319/4/gate/openstack-ansible-linters/f0c8c98/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/19/660319/4/gate/openstack-tox-docs/bd9ea0b/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":4},{"id":"7222522317e17898c5d30d0ad8add3291b121bfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 13:30:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cdd69640de995f4898d6bd475dd20be627a7fca2","revisions":{"3b9ff3127b4d2e6e0a8b5fcf7f27fe6c12286925":{"kind":"REWORK","_number":1,"created":"2019-05-21 09:42:32.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/660319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/19/660319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/1"}}},"commit":{"parents":[{"commit":"50167b448ff6180266fc53579ca310f5a75297ef","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/50167b448ff6180266fc53579ca310f5a75297ef"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:41:07.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:42:12.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nThis also makes sure leap 42.3 is not tested anymore.\n\nChange-Id: Ia2d252cfe4a7ef6d10369635a320c208262692d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/3b9ff3127b4d2e6e0a8b5fcf7f27fe6c12286925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/3b9ff3127b4d2e6e0a8b5fcf7f27fe6c12286925"}]},"branch":"refs/heads/master"},"f150e1406947824c3272c8f4d0a641e3c2e7bc2f":{"kind":"REWORK","_number":2,"created":"2019-05-21 09:56:01.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/660319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/19/660319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/2"}}},"commit":{"parents":[{"commit":"50167b448ff6180266fc53579ca310f5a75297ef","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/50167b448ff6180266fc53579ca310f5a75297ef"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:41:07.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:55:59.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nThis also makes sure leap 42.3 is not tested anymore.\n\nChange-Id: Ia2d252cfe4a7ef6d10369635a320c208262692d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f150e1406947824c3272c8f4d0a641e3c2e7bc2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f150e1406947824c3272c8f4d0a641e3c2e7bc2f"}]},"branch":"refs/heads/master"},"783cba13450e9181139970727f21576df242eefa":{"kind":"REWORK","_number":3,"created":"2019-05-22 07:20:05.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/660319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/19/660319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/3"}}},"commit":{"parents":[{"commit":"50167b448ff6180266fc53579ca310f5a75297ef","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/50167b448ff6180266fc53579ca310f5a75297ef"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:41:07.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-22 07:20:03.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nThis also makes sure leap 42.3 is not tested anymore.\n\nChange-Id: Ia2d252cfe4a7ef6d10369635a320c208262692d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/783cba13450e9181139970727f21576df242eefa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/783cba13450e9181139970727f21576df242eefa"}]},"branch":"refs/heads/master"},"cdd69640de995f4898d6bd475dd20be627a7fca2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-05-22 07:20:48.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/19/660319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/19/660319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/19/660319/4"}}},"commit":{"parents":[{"commit":"50167b448ff6180266fc53579ca310f5a75297ef","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/50167b448ff6180266fc53579ca310f5a75297ef"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:41:07.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-22 07:20:46.000000000","tz":0},"subject":"Make coverage consistent","message":"Make coverage consistent\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nThis also makes sure leap 42.3 is not tested anymore,\nand adds the linting jobs.\n\nChange-Id: Ia2d252cfe4a7ef6d10369635a320c208262692d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/cdd69640de995f4898d6bd475dd20be627a7fca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/cdd69640de995f4898d6bd475dd20be627a7fca2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
