)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~660320","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I287e8e46dc8548d6ebeea966ed10b9e2a2d97965","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"I287e8e46dc8548d6ebeea966ed10b9e2a2d97965","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-21 09:43:21.000000000","updated":"2019-05-22 23:30:21.000000000","submitted":"2019-05-22 23:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660320-1558567821141-7a2beb0f","meta_rev_id":"dbedf572179d221be2670b8a466833d4a4761f8d","_number":660320,"virtual_id_number":660320,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-22 20:48:48.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-22 23:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-22 14:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-22 17:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 14:55:37.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-22 17:53:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-22 20:48:48.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-22 23:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cdba8c3f76f9fbf6712cdfa3f9d860fa0ba4d8a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 09:43:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"226860c2c4457b2226873bcd83ad0e292f0cae0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 09:43:36.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-ansible-galera_client on branch master.  The\nerror was:\n\n  Job openstack-ansible-ssl-galera_server-opensuse-150 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-ansible-deploy-aio_metal-jobs\n        - openstack-ansible-deploy-aio_distro_metal-jobs\n        - check-requirements\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-ansible-ssl-galera_server-opensuse-150\n          - openstack-ansible-ssl-galera_server-ubuntu-xenial\n\n  in \"openstack/openstack-ansible-galera_client/zuul.d/project.yaml@master\", line 16, column 3","accounts_in_message":[],"_revision_number":1},{"id":"a892691d00dd9f89ba165919449752591d44f66d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 12:35:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4f314724d490dde30d1d826dced6f7ba9cb1c53b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 13:53:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/20/660320/2/check/openstack-ansible-deploy-aio_metal-debian-stable/9ea2d91/ : SUCCESS in 58m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/20/660320/2/check/openstack-ansible-deploy-aio_metal-centos-7/c6fb11a/ : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/20/660320/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/02165a3/ : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/20/660320/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f206647/ : SUCCESS in 50m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/20/660320/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/1d46c83/ : SUCCESS in 54m 30s\n- openstack-tox-docs http://logs.openstack.org/20/660320/2/check/openstack-tox-docs/dff6e5c/html/ : SUCCESS in 2m 29s\n- openstack-ansible-ssl-galera_server-opensuse-15 http://logs.openstack.org/20/660320/2/check/openstack-ansible-ssl-galera_server-opensuse-15/51ba176/ : FAILURE in 33m 11s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/20/660320/2/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/2cab8f2/ : SUCCESS in 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58991086cc2523820106ff06432e2da4676e205c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-22 13:37:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"403fa61780a92e25e56e0fc9966ffec2463a9193","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-22 14:55:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3c3c040f628cb24afa6656694482cd42367670dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 15:25:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-debian-stable/6fb3d4b/ : SUCCESS in 58m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-centos-7/7ed817e/ : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/941b7b0/ : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dadc9b1/ : SUCCESS in 1h 00m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/2588239/ : SUCCESS in 57m 46s\n- openstack-ansible-linters http://logs.openstack.org/20/660320/3/check/openstack-ansible-linters/f8b799a/ : SUCCESS in 6m 27s\n- openstack-tox-docs http://logs.openstack.org/20/660320/3/check/openstack-tox-docs/02dafff/html/ : SUCCESS in 2m 56s\n- openstack-ansible-ssl-galera_server-opensuse-15 http://logs.openstack.org/20/660320/3/check/openstack-ansible-ssl-galera_server-opensuse-15/a33e480/ : FAILURE in 22m 41s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/20/660320/3/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/d57207c/ : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"858e8c5c76565229387cd99af8e71c8dbb3e919d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-22 17:53:57.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1386e481f32ede666c2bb5f9988d4496772a1eb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 17:54:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ba6b863fac61fa2fe6a5f7b34b5124d9b74bc8d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 20:11:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable finger://ze05.openstack.org/ffd612ebdd0e40b98e52c9886f012b41 : RETRY_LIMIT in 39m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 finger://ze03.openstack.org/7c644665b6b645bc9123553026c83c99 : RETRY_LIMIT in 31m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-150 finger://ze02.openstack.org/fdb89d4fc0a6491bbbcb97b7b7ce33e8 : RETRY_LIMIT in 38m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic finger://ze12.openstack.org/91e6ef023b7540fab1f1eba4942e1793 : RETRY_LIMIT in 38m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 finger://ze04.openstack.org/85b2cec886274df483e95977d49d40db : RETRY_LIMIT in 28m 36s\n- openstack-ansible-linters http://logs.openstack.org/20/660320/3/gate/openstack-ansible-linters/b97008c/ : SUCCESS in 6m 05s\n- openstack-tox-docs http://logs.openstack.org/20/660320/3/gate/openstack-tox-docs/935f073/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"e5a63157ac61afface15ca77fd03de1706e6c456","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-22 20:48:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d0a5fc1e05c3b41b469faca637c6d16346396c84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 22:11:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-debian-stable/4dab52f/ : SUCCESS in 1h 01m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-centos-7/2dbe5fb/ : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/fc89520/ : SUCCESS in 1h 21m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d712eb1/ : SUCCESS in 55m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/20/660320/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b575675/ : SUCCESS in 58m 27s\n- openstack-ansible-linters http://logs.openstack.org/20/660320/3/check/openstack-ansible-linters/6be4201/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/20/660320/3/check/openstack-tox-docs/0b2b7ef/html/ : SUCCESS in 2m 24s\n- openstack-ansible-ssl-galera_server-opensuse-15 http://logs.openstack.org/20/660320/3/check/openstack-ansible-ssl-galera_server-opensuse-15/10892b0/ : FAILURE in 25m 41s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/20/660320/3/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/99c1d47/ : SUCCESS in 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49916f90baed8ed76ed371bc92081f329b88b285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 22:11:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"011735cf73aad8aa153183d303c2cadeacabe460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 23:30:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/20/660320/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/8075fe8/ : SUCCESS in 58m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/20/660320/3/gate/openstack-ansible-deploy-aio_metal-centos-7/afdbb52/ : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/20/660320/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/2c85d31/ : SUCCESS in 1h 06m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/20/660320/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e51d046/ : SUCCESS in 53m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/20/660320/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/f1abeba/ : SUCCESS in 1h 00m 32s\n- openstack-ansible-linters http://logs.openstack.org/20/660320/3/gate/openstack-ansible-linters/7f27e98/ : SUCCESS in 6m 46s\n- openstack-tox-docs http://logs.openstack.org/20/660320/3/gate/openstack-tox-docs/5225171/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"22806450c1fbc39861f0991f1fe4a5cbca67850e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 23:30:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7fa820fb6e41aeb26e131b9aaf72053f0254adca","revisions":{"b910892822a8da6f7a1e2f4b1bb7bf937bd2230d":{"kind":"REWORK","_number":1,"created":"2019-05-21 09:43:21.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/20/660320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/20/660320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/1"}}},"commit":{"parents":[{"commit":"3b9ff3127b4d2e6e0a8b5fcf7f27fe6c12286925","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/3b9ff3127b4d2e6e0a8b5fcf7f27fe6c12286925"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:42:53.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:42:53.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I287e8e46dc8548d6ebeea966ed10b9e2a2d97965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/b910892822a8da6f7a1e2f4b1bb7bf937bd2230d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/b910892822a8da6f7a1e2f4b1bb7bf937bd2230d"}]},"branch":"refs/heads/master"},"e62da60318cd37d77aed1bbc83126b3410ad9757":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-21 12:35:38.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/20/660320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/20/660320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/2"}}},"commit":{"parents":[{"commit":"f150e1406947824c3272c8f4d0a641e3c2e7bc2f","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f150e1406947824c3272c8f4d0a641e3c2e7bc2f"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:42:53.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 12:35:38.000000000","tz":0},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I287e8e46dc8548d6ebeea966ed10b9e2a2d97965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/e62da60318cd37d77aed1bbc83126b3410ad9757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/e62da60318cd37d77aed1bbc83126b3410ad9757"}]},"branch":"refs/heads/master"},"7fa820fb6e41aeb26e131b9aaf72053f0254adca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-22 13:37:14.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/20/660320/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/20/660320/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/20/660320/3"}}},"commit":{"parents":[{"commit":"cdd69640de995f4898d6bd475dd20be627a7fca2","subject":"Make coverage consistent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/cdd69640de995f4898d6bd475dd20be627a7fca2"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 09:42:53.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-22 13:37:14.000000000","tz":0},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I287e8e46dc8548d6ebeea966ed10b9e2a2d97965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/7fa820fb6e41aeb26e131b9aaf72053f0254adca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/7fa820fb6e41aeb26e131b9aaf72053f0254adca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
