)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~664072","triplet_id":"openstack%2Fopenstack-ansible-galera_client~stable%2Fqueens~I39ba1d72ec21848b172d70e9f74acde86248044a","project":"openstack/openstack-ansible-galera_client","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I39ba1d72ec21848b172d70e9f74acde86248044a","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:20:29.000000000","updated":"2019-06-08 17:32:36.000000000","submitted":"2019-06-08 17:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664072-1560015156511-22ee9161","meta_rev_id":"f2de3ccffb77ad9d8965a503751ad37f67ba1c16","_number":664072,"virtual_id_number":664072,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-08 17:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-06-08 17:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-06-08 17:19:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 17:19:19.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-08 17:32:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"774d8e8cecb4a5b904d59c7a5fe96deffeb71940","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"674fe1c39caa5b52eb7a1aa900489ba6712d4137","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:48:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/664072/1/check/build-openstack-sphinx-docs/55e7224/html/ : SUCCESS in 3m 05s\n- openstack-ansible-linters http://logs.openstack.org/72/664072/1/check/openstack-ansible-linters/42c1008/ : SUCCESS in 5m 47s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/664072/1/check/openstack-ansible-functional-centos-7/cf63ea1/ : SUCCESS in 7m 19s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/72/664072/1/check/openstack-ansible-functional-opensuse-423/4c5a2b7/ : SUCCESS in 6m 41s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/72/664072/1/check/openstack-ansible-functional-ubuntu-xenial/0ee91b3/ : SUCCESS in 5m 26s\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/72/664072/1/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/211a7c3/ : SUCCESS in 18m 35s (non-voting)\n- openstack-ansible-functional-galera_server-ubuntu-xenial http://logs.openstack.org/72/664072/1/check/openstack-ansible-functional-galera_server-ubuntu-xenial/008045a/ : SUCCESS in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a8937841c08a94807e5f6edd6ea0af6ca5bb977","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-08 17:19:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94735164e75be1217bba47750924237cd3b245a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:19:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36c21c76eb37ba6faf8de1ec140dae46e698adbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:32:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0be1ecffc03352cf3b170a7facff1d07570db8ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 17:32:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/664072/1/gate/build-openstack-sphinx-docs/6437bd6/html/ : SUCCESS in 3m 07s\n- openstack-ansible-linters http://logs.openstack.org/72/664072/1/gate/openstack-ansible-linters/c969b36/ : SUCCESS in 5m 04s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/664072/1/gate/openstack-ansible-functional-centos-7/e8d2337/ : SUCCESS in 7m 23s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/72/664072/1/gate/openstack-ansible-functional-opensuse-423/12c1290/ : SUCCESS in 5m 39s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/72/664072/1/gate/openstack-ansible-functional-ubuntu-xenial/0836508/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40a743c72d5e91be3354fdc55662b408b426ffd3","revisions":{"40a743c72d5e91be3354fdc55662b408b426ffd3":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:20:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/664072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/72/664072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/72/664072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/72/664072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/72/664072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/72/664072/1"}}},"commit":{"parents":[{"commit":"100f5e4525975c3173e47f4bfc26ab8c9e47630c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/100f5e4525975c3173e47f4bfc26ab8c9e47630c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:20:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:20:25.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I39ba1d72ec21848b172d70e9f74acde86248044a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/40a743c72d5e91be3354fdc55662b408b426ffd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/40a743c72d5e91be3354fdc55662b408b426ffd3"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
