)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~664285","triplet_id":"openstack%2Fopenstack-ansible-galera_client~stable%2Frocky~Ia1516c49855b69bfdfcfc2fb25e81208b11b0ac1","project":"openstack/openstack-ansible-galera_client","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia1516c49855b69bfdfcfc2fb25e81208b11b0ac1","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:43:14.000000000","updated":"2019-06-10 17:30:25.000000000","submitted":"2019-06-10 17:30:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664285-1560187825537-d9eb55b5","meta_rev_id":"e3d91b0b68f3b12da7c27307867b498081e12b49","_number":664285,"virtual_id_number":664285,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:10:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:10:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e56d41dfc118b2fd9df5a3812a4193f3655213e1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:43:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39f5af770183b5dc2d55973fb08f249064f4026a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:17:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/664285/1/check/openstack-ansible-linters/7744320/ : SUCCESS in 6m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-centos-7/5df4d2e/ : SUCCESS in 12m 03s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-opensuse-423/302cc40/ : SUCCESS in 7m 12s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-opensuse-150/52defbf/ : FAILURE in 6m 01s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-ubuntu-bionic/03bd5c7/ : SUCCESS in 4m 55s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-ubuntu-xenial/0aa78d0/ : SUCCESS in 4m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/664285/1/check/build-openstack-sphinx-docs/d410cd1/html/ : SUCCESS in 4m 17s\n- openstack-ansible-ssl-galera_server-opensuse-423 http://logs.openstack.org/85/664285/1/check/openstack-ansible-ssl-galera_server-opensuse-423/68b1749/ : FAILURE in 8m 04s (non-voting)\n- openstack-ansible-ssl-galera_server-ubuntu-xenial http://logs.openstack.org/85/664285/1/check/openstack-ansible-ssl-galera_server-ubuntu-xenial/799798f/ : SUCCESS in 18m 57s (non-voting)\n- openstack-ansible-functional-galera_server-opensuse-423 http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-galera_server-opensuse-423/702f199/ : FAILURE in 7m 57s (non-voting)\n- openstack-ansible-functional-galera_server-ubuntu-xenial http://logs.openstack.org/85/664285/1/check/openstack-ansible-functional-galera_server-ubuntu-xenial/02c068e/ : SUCCESS in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"595385c77ab68654373c4ddd6e2b851b2cd3ef21","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:10:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48961d49e72ab08bf018f8cf5612e1073031457a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:10:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00acf910833b79b633c1024b51b9af67a5720f41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:30:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54cbb51efd89e855687610f23c64f6100d6a832d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:30:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/664285/1/gate/openstack-ansible-linters/fa13979/ : SUCCESS in 6m 10s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/85/664285/1/gate/openstack-ansible-functional-centos-7/97b8f55/ : SUCCESS in 10m 44s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/85/664285/1/gate/openstack-ansible-functional-opensuse-423/391f602/ : SUCCESS in 6m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/85/664285/1/gate/openstack-ansible-functional-ubuntu-bionic/abe2424/ : SUCCESS in 9m 55s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/85/664285/1/gate/openstack-ansible-functional-ubuntu-xenial/a0be8e5/ : SUCCESS in 5m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/85/664285/1/gate/build-openstack-sphinx-docs/0e39522/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71f3d9515efd5833f95f8c05a0a01886149b4dbd","revisions":{"71f3d9515efd5833f95f8c05a0a01886149b4dbd":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:43:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/664285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/85/664285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/85/664285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/85/664285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/85/664285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/85/664285/1"}}},"commit":{"parents":[{"commit":"28d89bee42b5a91f590591737c2aedd307272fad","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/28d89bee42b5a91f590591737c2aedd307272fad"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:13.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia1516c49855b69bfdfcfc2fb25e81208b11b0ac1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/71f3d9515efd5833f95f8c05a0a01886149b4dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/71f3d9515efd5833f95f8c05a0a01886149b4dbd"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
