)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~664286","triplet_id":"openstack%2Fopenstack-ansible-galera_client~stable%2Fpike~I205224c83454968816bf09ef1a62f430e52ca481","project":"openstack/openstack-ansible-galera_client","branch":"stable/pike","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I205224c83454968816bf09ef1a62f430e52ca481","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:43:16.000000000","updated":"2019-06-10 17:51:23.000000000","submitted":"2019-06-10 17:51:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664286-1560189083686-e5d355c6","meta_rev_id":"e412f08b00f38e0c7695505c313d1f9f7e1dc05b","_number":664286,"virtual_id_number":664286,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:31:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:51:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e7f49ef8555b6a85adf63037b5734ff1a313cda","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:43:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1abf1ae08de6ee1daabe556d9d347bcde108c1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:34:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/86/664286/1/check/build-openstack-sphinx-docs/14773bf/html/ : SUCCESS in 3m 42s\n- openstack-ansible-linters http://logs.openstack.org/86/664286/1/check/openstack-ansible-linters/880cb43/ : SUCCESS in 5m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/86/664286/1/check/openstack-ansible-functional-centos-7/6c3e04e/ : SUCCESS in 9m 54s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/86/664286/1/check/openstack-ansible-functional-opensuse-423/9c9ee80/ : SUCCESS in 6m 33s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/86/664286/1/check/openstack-ansible-functional-ubuntu-xenial/e669750/ : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"afa13dedb77a2ba26319f9f6b7a34fcdf1bdef52","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:31:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9b7bb6f3509109819933d8e36f8e970ef3c290a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:31:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd294223489af023d4365078c7816d8da1e62893","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:51:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b9e4354600ba91c02db5ebe6a1940ec1ce2e5d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:51:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/86/664286/1/gate/build-openstack-sphinx-docs/3865eb1/html/ : SUCCESS in 2m 59s\n- openstack-ansible-linters http://logs.openstack.org/86/664286/1/gate/openstack-ansible-linters/ac6f7b4/ : SUCCESS in 4m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/86/664286/1/gate/openstack-ansible-functional-centos-7/c1d8e80/ : SUCCESS in 11m 29s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/86/664286/1/gate/openstack-ansible-functional-opensuse-423/c14a5a0/ : SUCCESS in 6m 40s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/86/664286/1/gate/openstack-ansible-functional-ubuntu-xenial/711141b/ : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e346cad7ed8a99bc07e65e8d98df24b3c7570e8b","revisions":{"e346cad7ed8a99bc07e65e8d98df24b3c7570e8b":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:43:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/664286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/86/664286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/86/664286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/86/664286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/86/664286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/86/664286/1"}}},"commit":{"parents":[{"commit":"c08ea49e7d31e5292e27e8b020e41458194f02b5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/c08ea49e7d31e5292e27e8b020e41458194f02b5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:14.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I205224c83454968816bf09ef1a62f430e52ca481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/e346cad7ed8a99bc07e65e8d98df24b3c7570e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/e346cad7ed8a99bc07e65e8d98df24b3c7570e8b"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
