)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~682229","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I365d69f2c4e692101806cde286c504d77f1a40eb","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"osa-focal","hashtags":[],"change_id":"I365d69f2c4e692101806cde286c504d77f1a40eb","subject":"Bump galera version to 10.5.2","status":"ABANDONED","created":"2019-09-14 20:48:59.000000000","updated":"2023-02-01 15:38:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc7e4e146483eb12ffb9ef98ab8205e2e68f79a6","_number":682229,"virtual_id_number":682229,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-28 08:55:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-05-27 17:20:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 08:55:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93e913d52434dfbb18fc7644a96c55648d2ca377","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-14 20:48:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9cc73ed511694302c342e2938f63031e50d51a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 21:26:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9ea87f7176b34fc99899fa4e916d9c22 : FAILURE in 21m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/20ef4141354f452dbb10e04a338dc9b0 : FAILURE in 26m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/425757ec0a2b41f98a4c56ca5e3369dc : FAILURE in 23m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ef8f9b3156b45969e53d4e9937b571f : FAILURE in 19m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8c4e72eea177446b8f87dd9845a60bfe : FAILURE in 19m 13s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/596a854d16d648de81212dd9b8379ec5 : FAILURE in 18m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd4bb4b799aa42f19d6623cff944c55f : FAILURE in 15m 28s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/920c67d92b33480cb56567d4f51c3b7c : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84c3bd5b16e40c2b84e345d401dfb66 : SUCCESS in 2m 59s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/7768652db0884f5aafe8854387ea055a : RETRY_LIMIT in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5519b0c83fc7c9c21aeb3ad52f01d2befd6ae59","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-18 16:46:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11177cfc7764703c8d37c3fbc7acd85f737f74e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 17:21:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e10b8ceee3594a5084b399396d116ff7 : FAILURE in 18m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c6e2e7e823494fffa62f68443f796ca8 : FAILURE in 22m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1f60bac9c8ee49f5b963ccdc17b2725f : FAILURE in 23m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e70661f0bd5047cdb75cda7939691c9e : FAILURE in 20m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a14e627be0cf455ea4ce3a4a967b812e : FAILURE in 20m 16s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/29c9b464cd3d452a837085f0ecf16b4d : FAILURE in 22m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba824a4498af49e2b3df31e478f5f514 : FAILURE in 19m 31s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/65623b13be6549c49075839d535f258a : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c823bf464a7a4cd8ba1ade453c551b25 : SUCCESS in 3m 26s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/520a9150a0df44a79c072a8164e436ba : FAILURE in 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae14437094ef3f3c3dd78eceb00ba7017013566f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-31 18:01:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3ac3cd45c9849c42c5a5d43dd95b04d3a24a38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 18:01:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1345fc89f7f2460b0947bffe9e00ac41c135e9b4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-27 17:38:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"666d223293499cea1f24c0347e778f54149281ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:38:42.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"e638f928d8f7dc047d3835d9c3da28d2cfd6fcd2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-27 20:33:26.000000000","message":"Topic set to osa-focal","accounts_in_message":[],"_revision_number":2},{"id":"0803f94b09091924594623c75cb12931fbd902b6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-28 06:06:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50801761fecb496bc24680dbea4d0aa27a2557f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-28 08:08:30.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4d45120ddd5b6db13e50330574a473ee1f9093be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:55:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b58341ea35e740a8b2f204e8c26fb94d : FAILURE in 24m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1f3b6b3393344b04890c77dc6cd23eb5 : FAILURE in 31m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a70afbe48bc04d6f9cc9784c6e62b139 : FAILURE in 26m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08cac799f19f4be493391f59d838c5ff : FAILURE in 20m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a916381971a04e94acf2283fb1121c28 : EXCEPTION (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa2e628781624d9a93baef629405259f : EXCEPTION\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4b9ee0683c145d9b2661dc524f84945 : FAILURE in 23m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cffb595c340149ada28f392f6d66a532 : FAILURE in 19m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e22613660a649c49711dc45d54db439 : FAILURE in 17m 17s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1b91c4a80965403fab5221b291d17106 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76af865c559648428faee5399fdf7003 : SUCCESS in 13m 23s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/8a5bbf1e83b0478da9ac50fe9aba4b84 : FAILURE in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bc127c5cd97ef904b0dd1ae24b888418d06579d3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-27 17:20:01.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"dc7e4e146483eb12ffb9ef98ab8205e2e68f79a6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-02-01 15:38:43.000000000","message":"Abandoned\n\nHugely out of date now","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"026b15ec47072eb461babc82a062949a6c3b6f61","revisions":{"018224eb26fbd15a0c38d12ad8dec6f2466c8025":{"kind":"REWORK","_number":1,"created":"2019-09-14 20:48:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/682229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/29/682229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/1"}}},"commit":{"parents":[{"commit":"315d342692d0af565d68d294845049ed56195d10","subject":"Update invalid link for README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/315d342692d0af565d68d294845049ed56195d10"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-14 20:39:12.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-14 20:48:34.000000000","tz":60},"subject":"Bump galera version to 10.4.8","message":"Bump galera version to 10.4.8\n\nChange-Id: I365d69f2c4e692101806cde286c504d77f1a40eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/018224eb26fbd15a0c38d12ad8dec6f2466c8025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/018224eb26fbd15a0c38d12ad8dec6f2466c8025"}]},"branch":"refs/heads/master"},"63f8e92c343d2a903f85339b0aec12ccef8813ab":{"kind":"REWORK","_number":2,"created":"2020-04-27 17:38:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/682229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/29/682229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/2"}}},"commit":{"parents":[{"commit":"315d342692d0af565d68d294845049ed56195d10","subject":"Update invalid link for README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/315d342692d0af565d68d294845049ed56195d10"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-14 20:39:12.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-27 17:38:10.000000000","tz":0},"subject":"Bump galera version to 10.5.2","message":"Bump galera version to 10.5.2\n\nChange-Id: I365d69f2c4e692101806cde286c504d77f1a40eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/63f8e92c343d2a903f85339b0aec12ccef8813ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/63f8e92c343d2a903f85339b0aec12ccef8813ab"}]},"branch":"refs/heads/master"},"9997d14238adc47dffe7e03401e57aeea0723105":{"kind":"REWORK","_number":3,"created":"2020-04-28 06:06:53.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/682229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/29/682229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/3"}}},"commit":{"parents":[{"commit":"6d6ea45e0e3ee072357d383f708277a600cbf4cd","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/6d6ea45e0e3ee072357d383f708277a600cbf4cd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-14 20:39:12.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-28 06:06:48.000000000","tz":60},"subject":"Bump galera version to 10.5.2","message":"Bump galera version to 10.5.2\n\nChange-Id: I365d69f2c4e692101806cde286c504d77f1a40eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/9997d14238adc47dffe7e03401e57aeea0723105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/9997d14238adc47dffe7e03401e57aeea0723105"}]},"branch":"refs/heads/master"},"026b15ec47072eb461babc82a062949a6c3b6f61":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-28 08:08:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/682229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/29/682229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/29/682229/4"}}},"commit":{"parents":[{"commit":"6d6ea45e0e3ee072357d383f708277a600cbf4cd","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/6d6ea45e0e3ee072357d383f708277a600cbf4cd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-14 20:39:12.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-28 08:08:27.000000000","tz":0},"subject":"Bump galera version to 10.5.2","message":"Bump galera version to 10.5.2\n\nDepends-On: https://review.opendev.org/723276\nChange-Id: I365d69f2c4e692101806cde286c504d77f1a40eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/026b15ec47072eb461babc82a062949a6c3b6f61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/026b15ec47072eb461babc82a062949a6c3b6f61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
