)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~689010","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I791eb263daca896aa4538f38938b43f2a4a75b5c","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"opensuse-150","hashtags":[],"change_id":"I791eb263daca896aa4538f38938b43f2a4a75b5c","subject":"Switch to opensuse-15 nodeset","status":"MERGED","created":"2019-10-16 19:19:20.000000000","updated":"2019-10-18 18:57:34.000000000","submitted":"2019-10-18 18:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689010-1571424877109-ed58b23b","meta_rev_id":"356742adec96d9edf4098ad6e861fa7ce1005df9","_number":689010,"virtual_id_number":689010,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2019-10-18 13:45:46.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 18:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-10-18 07:52:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 17:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"date":"2019-10-18 06:39:26.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 17:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 06:39:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-10-18 07:52:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-18 17:17:08.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-18 18:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22fed2d0bc58016371a1e1eb9d2b74bff2d980a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-16 19:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1961b44acd7f803e3df884c709b4843c6de093e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 21:51:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/101add555167444f8097872cfc69c75f : SUCCESS in 1h 23m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f2eeebeed5d84949898531c57b35ee0c : SUCCESS in 1h 46m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9df9d20a6a8b4e26ace6641bf9dd1c27 : FAILURE in 52m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51fb3aa93b3844ed9f7510229561d88c : SUCCESS in 1h 14m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/38b7a785da024094bf647690c3a3f887 : SUCCESS in 1h 08m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba8cebd2295547ab8a65286ae1225d11 : FAILURE in 38m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92a24eb7bf514df88c18c0b9849fc1c7 : SUCCESS in 53m 35s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2a15a021439e457093745534875fa5f8 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d51df72f0d4e63a924576d06ac7984 : SUCCESS in 4m 37s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/5afee4ce927c4cff94358eee15e7c190 : FAILURE in 16m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"320cad32a1f9a04e5aba7e63b5f9a71d4198adbc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-18 04:49:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a97f65447b87ed66bcf1be3de13576ba9579b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 06:12:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9d88d22a9e60499bbb6dc178d08bd1e7 : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aff7a907ec554451b6c9ac2292a19efb : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8197327721174369b2fe7bab013d3dab : FAILURE in 46m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5280537255ac45619a833dd6dbbd7e15 : SUCCESS in 1h 04m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5a26f57fbbc549a98fc23c96b30a81bd : SUCCESS in 1h 06m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/baea73fcc0554898b608b6c5475ddefe : FAILURE in 34m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c502371eb644291862ec48e4fc1d947 : SUCCESS in 50m 16s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8350e5b3aa6048bba823b5280b2256cc : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae26f2c2fcd4d0dbc033c1119c16f98 : SUCCESS in 4m 36s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/10795146249741168d30a951eac0f777 : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"467fee320b90f42afdeafb63e692fad0e41f3e9c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-10-18 06:39:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f47b5fd75c5e5891554718541c4c249067a4b66c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 07:52:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"50571cf789dc4cd5a68e4f71b080f4dc029419b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 08:29:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c7bfe47f47a244ab93d32231c1b9471e : SUCCESS in 1h 12m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5a21c09d854c4c4ebf32fb8e848e2552 : SUCCESS in 1h 31m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4fdd4f98c4264bd2b3afe5d126556786 : FAILURE in 52m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99ee9a4748e44c0897d78167e3af0c59 : SUCCESS in 1h 17m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/349df7e7708c463ca938ae3a2d6f84af : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/040e48ea95884ad680fdb2f273716fbe : FAILURE in 50m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cee254d6079409eacaa0131e15a5892 : SUCCESS in 1h 01m 41s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/da081562395a45109ef08a448e7d13f7 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee321f05b26449c84c61f262d6251ac : SUCCESS in 7m 19s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/fe519f5022574f948fffe55263a2b2e8 : FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9cc7ba46d92191cea41855655dbf8bb84d8ff05f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-18 13:45:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d87c97374673c842e8a11eda4aabc746163636d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:23:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/02697aa6edde4b1bbbeda266fbf84fb5 : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/727632fda6b4458fb3ec867f08d22f00 : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b009406edcb4453993b0b6f98488a224 : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83cd33f9f6bd4659ae555f8227aee7db : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b11433f94564e4d99c605f8de5683c1 : SUCCESS in 1h 11m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/64d44811cc4148caa0e3eea23c81a996 : FAILURE in 37m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b755312397a74770a123af892d3f87c9 : SUCCESS in 54m 32s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e4503ddf25f94608b18a94b805904c05 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9e7b4a8eee4a1480db1c7df26d2891 : SUCCESS in 4m 47s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/6c616ee335e446b09981537f0ee5ccd9 : FAILURE in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9d833ba95e2fb9382e2abae1fc606550a8e2537","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-18 17:17:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"038348ad5e64c3b14a64e05b8cc36cf939ada9ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 17:17:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fea9b09d44af40ae70484dddd635401cff7caa52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:54:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3446a1567c47410288d95cc80e2709e5 : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0735538cd4fd488f9f0b40588a6a9d67 : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9b22376cce7a46cab012a51093400231 : SUCCESS in 1h 21m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0982641e66b1409387da8edc094a258e : SUCCESS in 1h 07m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7670dc01a117447e909b7b3e9a656852 : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d449185d9e74ebcbf3b01883cbbc16a : SUCCESS in 47m 00s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ebffa6b773534287a7694190f72758de : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d7b08082d84e4b9c20ccce8ae2ec0b : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8bd6a9f87405d244eae46af395b7babf842b6048","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:54:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b9a284d29cab694481298bc65ca548d14c1d55e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:57:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172b59e9e8024d5d9cca7a56fb3f7940 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dd4cadbdea8ea0602927571ae9d42aebaa32456f","revisions":{"7ed8d3556400869dc483f8a9f804987ccc4e22de":{"kind":"REWORK","_number":1,"created":"2019-10-16 19:19:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/689010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/10/689010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/1"}}},"commit":{"parents":[{"commit":"f0b69f594e0dcf16a843bd2cd5e1d658ea2ff0ac","subject":"Bump galera version to 10.3.18","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f0b69f594e0dcf16a843bd2cd5e1d658ea2ff0ac"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:19:10.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:19:10.000000000","tz":120},"subject":"Switch to opensuse-15 nodeset","message":"Switch to opensuse-15 nodeset\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nChange-Id: I791eb263daca896aa4538f38938b43f2a4a75b5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/7ed8d3556400869dc483f8a9f804987ccc4e22de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/7ed8d3556400869dc483f8a9f804987ccc4e22de"}]},"branch":"refs/heads/master"},"dd4cadbdea8ea0602927571ae9d42aebaa32456f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-18 06:39:26.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/10/689010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/10/689010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/10/689010/2"}}},"commit":{"parents":[{"commit":"f0b69f594e0dcf16a843bd2cd5e1d658ea2ff0ac","subject":"Bump galera version to 10.3.18","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f0b69f594e0dcf16a843bd2cd5e1d658ea2ff0ac"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-16 19:19:10.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-10-18 06:39:14.000000000","tz":120},"subject":"Switch to opensuse-15 nodeset","message":"Switch to opensuse-15 nodeset\n\nopensuse-150 nodeset is referring to openSUSE 15.0, which is still in\nmaintenance but openSUSE 15.1 has been released already. \"opensuse-15\"\nis going to refer to the \"latest openSUSE 15.x\" build released and\nworking for OpenStack going forward, so add this nodeset and use\nit by default going forward.\n\nDepends-On: https://review.opendev.org/#/c/689323/\nChange-Id: I791eb263daca896aa4538f38938b43f2a4a75b5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/dd4cadbdea8ea0602927571ae9d42aebaa32456f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/dd4cadbdea8ea0602927571ae9d42aebaa32456f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
