)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~690658","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I4a931463119d0a48c83a9217b58868353170e5f3","project":"openstack/openstack-ansible-galera_client","branch":"master","hashtags":[],"change_id":"I4a931463119d0a48c83a9217b58868353170e5f3","subject":"tox: Keeping going with docs","status":"MERGED","created":"2019-10-23 17:19:07.000000000","updated":"2019-11-21 16:45:42.000000000","submitted":"2019-11-21 16:43:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690658-1574354616986-64a7fc44","meta_rev_id":"f0d2ed21c99a20e66c4a76dfea05676a6056ff44","_number":690658,"virtual_id_number":690658,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 16:43:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-11-14 15:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2019-11-12 10:41:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-11 09:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":2,"date":"2019-11-18 10:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":1,"date":"2019-11-21 16:30:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 09:07:42.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-11-12 10:41:46.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-14 15:12:24.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-11-21 16:30:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-21 16:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d87702a0795965812c4ade80c5a68506d245766","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-23 17:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fb984c4b37ed0eb2af4cf92373616a87a1b3d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 17:42:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/274e185d57fd44b785d1bd7e3f09f257 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9f68aea40144f3a97bbc81c1c66191 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e80b731f8a6047d390af9345cb994cc5 : SUCCESS in 2m 52s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/af0a36cb29e44d39907230d50c55b5c8 : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e357845b18649a11afb8093765e884e5e3a84eb9","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-23 17:44:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"593fe68c51760cf2fa3b9777f67f5a4297d20d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:03:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5d725548f3c749b48e8b2f58d1447b47 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393db7d3a4ad40169e06b16da2fc6afd : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/275291cafe5146659c4526faec6d2da1 : SUCCESS in 4m 32s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/d77863e93f9f4df4b2f13b65a0569ff2 : FAILURE in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e07cc1f9829d8f7b1d3f59167b9077e34bbadd91","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-11-11 09:07:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0efb14c36436c3ecb2c11f0d3f56dd1e81ac3e19","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-12 10:41:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ca2aa06958f98f1d8b305d9e9065ac6490f4002e","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-11-14 15:12:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3dbf56bbc452f1a6e5fa7f4e857c494fd1ed8879","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 10:14:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"58777da338871c73ed33038903bfc9f5ca2632c2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 16:30:30.000000000","message":"Removed Workflow+1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"59eae79d695501e57823f7eb134b942c4f834282","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 16:30:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa72c10744e042a35e7847e1e6cb467733b2cc84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:30:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4023f4587b6b82c524b1c5148a27988b59f70e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:43:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b1151415e0bcabab184a63bddb3fe3e7c85b077e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:43:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e552e928c1ab44bbb71300d3396af1f0 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b7c5a59bae4147ba739d6c14eaa048 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/205af3100f96458d89d9778932034adc : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f0d2ed21c99a20e66c4a76dfea05676a6056ff44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:45:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a183a0cb086490bb4a3023fd5b2f940 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/904e786a3df1445fa920acd95e2bcba6 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48ac69cd99e1ffd85935b5be9692831c409602b4","revisions":{"5c5b04b36abe796355ba70aae09e008b4a84591b":{"kind":"REWORK","_number":1,"created":"2019-10-23 17:19:07.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/58/690658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/58/690658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/1"}}},"commit":{"parents":[{"commit":"c0f11db478985619cab26b8a6b3193cb86b7fcaa","subject":"Merge \"Switch to opensuse-15 nodeset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/c0f11db478985619cab26b8a6b3193cb86b7fcaa"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:18:34.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:18:48.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: I4a931463119d0a48c83a9217b58868353170e5f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/5c5b04b36abe796355ba70aae09e008b4a84591b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/5c5b04b36abe796355ba70aae09e008b4a84591b"}]},"branch":"refs/heads/master"},"48ac69cd99e1ffd85935b5be9692831c409602b4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-23 17:44:19.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/58/690658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/58/690658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/58/690658/2"}}},"commit":{"parents":[{"commit":"c0f11db478985619cab26b8a6b3193cb86b7fcaa","subject":"Merge \"Switch to opensuse-15 nodeset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/c0f11db478985619cab26b8a6b3193cb86b7fcaa"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:18:34.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-10-23 17:44:17.000000000","tz":0},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\n1. Sphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n2. Remove the unnecessary blank line for tox.ini\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nDepends-On: Ie837d376bb1a1d93367369d25140dd8a9d5ac14e\nChange-Id: I4a931463119d0a48c83a9217b58868353170e5f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/48ac69cd99e1ffd85935b5be9692831c409602b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/48ac69cd99e1ffd85935b5be9692831c409602b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
