)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~729670","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I0d6247312e22c365c1e449ee5a18a8765789a7f4","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"pygments_style","hashtags":[],"change_id":"I0d6247312e22c365c1e449ee5a18a8765789a7f4","subject":"Update docstheme for style","status":"MERGED","created":"2020-05-20 16:44:35.000000000","updated":"2020-05-20 17:45:09.000000000","submitted":"2020-05-20 17:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729670-1589996518294-e2f5466d","meta_rev_id":"5ca4305ca09d6530a22e1ccf2ebf0cebe8f6de16","_number":729670,"virtual_id_number":729670,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 17:41:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2020-05-20 16:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-20 17:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-20 17:27:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 16:46:45.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2020-05-20 17:27:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-20 17:41:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ce50f9c076eeec26dfd6f18be02c33c37c42190","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 16:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e8f567a4097968874fec398a4ba2a032a53bcbf","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-05-20 16:46:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e36badbeabd3c023c03f5e29805960cc80e687b","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2020-05-20 16:47:31.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"4de67926128cddca136ca8af435326988d5167be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d3bcac15b174a7e8278498c09f10b50 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd3b2089aa2249a5994315af7084e706 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/737537a8678248eea5e4a870ed8deeed : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d47d0fa5e5b2a794372da983b831c92b5d993d05","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 17:27:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8be3833714f89e7bbc4443cccf79746da59a4204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:27:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4c8915bb965393ae5ace008244dd0fb3c937140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:41:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/51007cbcce1a4b3ea95c0ba65d18720a : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7ea72b1262429fbd8b449644f2e114 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eeaef7a74454fb9b9c194412928e134 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e2f59f7a1fa8b45586484565a78e75f507532bee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:41:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe29fceb349e9c8e18f6d0a3ca7e027645268d2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:45:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc98e2d8d01848b2a59620db01491e80 : SUCCESS in 2m 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9015b1890e5d456595527b98a54dd4da : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70d7bc30db18d0b8061c43c2199a426ee2996519","revisions":{"70d7bc30db18d0b8061c43c2199a426ee2996519":{"kind":"REWORK","_number":1,"created":"2020-05-20 16:44:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/70/729670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/70/729670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/70/729670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/70/729670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/70/729670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/70/729670/1"}}},"commit":{"parents":[{"commit":"b96e5eaa22245d39cfc73188f24e8ad7bffc8612","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/b96e5eaa22245d39cfc73188f24e8ad7bffc8612"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:44:30.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:44:30.000000000","tz":120},"subject":"Update docstheme for style","message":"Update docstheme for style\n\nNew version of openstackdocstheme (Victoria+) respects pygments_style.\nSince this repo is using now Victoria (master) requirements but has\nnot branched for Ussuri yet, it uses the new version.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nChange-Id: I0d6247312e22c365c1e449ee5a18a8765789a7f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/70d7bc30db18d0b8061c43c2199a426ee2996519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/70d7bc30db18d0b8061c43c2199a426ee2996519"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
