)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~733209","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~Iecdabf6ad83d4f75977dae880be52cf16474eb0a","project":"openstack/openstack-ansible-galera_client","branch":"master","topic":"reno-openstackdocstheme","hashtags":[],"change_id":"Iecdabf6ad83d4f75977dae880be52cf16474eb0a","subject":"Switch to newer openstackdocstheme and reno versions","status":"ABANDONED","created":"2020-06-03 16:22:36.000000000","updated":"2020-06-03 17:51:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1adf59520c5ff9facb7c6ff4b7155d17134e5e6e","_number":733209,"virtual_id_number":733209,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-03 16:35:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2020-06-03 17:14:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 16:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-03 17:14:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"c029f7e4bde1f40c3d6c213d1f7353c3e796b14a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 16:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18e9980ba3b48530e0cb9a95f5c0a95fae2ba746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 16:35:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/38b2c44e177343baa70576d243ab09ae : SUCCESS in 4m 05s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/63a0a2bfe92546bf815266d491c14fb2 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2b393c99034bf19367c2d431813f2d : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01be53dc86f3454f90950a4bf87ff7d8 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"39368c5805aa13055009390b6903f681feff8713","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-03 17:14:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb737eae49d98a9dac824e61e377678eefe36c92","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 17:51:30.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3ca178d20f38910f11ce41136c9fa900a36756f","revisions":{"a3ca178d20f38910f11ce41136c9fa900a36756f":{"kind":"REWORK","_number":1,"created":"2020-06-03 16:22:36.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/733209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/09/733209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/09/733209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/09/733209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/09/733209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/09/733209/1"}}},"commit":{"parents":[{"commit":"1403bc919af12b58d990bbf0948c3927b4ce3c1e","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/1403bc919af12b58d990bbf0948c3927b4ce3c1e"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-03 16:22:08.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-03 16:22:25.000000000","tz":120},"subject":"Switch to newer openstackdocstheme and reno versions","message":"Switch to newer openstackdocstheme and reno versions\n\nSwitch to openstackdocstheme 2.2.1 and reno 3.1.0 versions. Using\nthese versions will allow especially:\n* Linking from HTML to PDF document\n* Allow parallel building of documents\n* Fix some rendering problems\n\nUpdate Sphinx version as well.\n\nSet openstackdocs_pdf_link to link to PDF file. Note that\nthe link to the published document only works on docs.openstack.org\nwhere the PDF file is placed in the top-level html directory. The\nsite-preview places the PDF in a pdf directory.\n\nopenstackdocstheme renames some variables, so follow the renames\nbefore the next release removes them. A couple of variables are also\nnot needed anymore, remove them.\n\nSee also\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/014971.html\n\nChange-Id: Iecdabf6ad83d4f75977dae880be52cf16474eb0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/a3ca178d20f38910f11ce41136c9fa900a36756f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/a3ca178d20f38910f11ce41136c9fa900a36756f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
