)]}'
{"id":"openstack%2Fopenstack-ansible-galera_client~762041","triplet_id":"openstack%2Fopenstack-ansible-galera_client~master~I0d712c73e870e01ed5683bc6f31de5b3100de622","project":"openstack/openstack-ansible-galera_client","branch":"master","hashtags":[],"change_id":"I0d712c73e870e01ed5683bc6f31de5b3100de622","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"MERGED","created":"2020-11-10 05:11:27.000000000","updated":"2020-11-10 11:59:41.000000000","submitted":"2020-11-10 11:57:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762041-1605009477418-0535b842","meta_rev_id":"9b74cb09e417dff7f7086bb82ff9195cd4b29c77","_number":762041,"virtual_id_number":762041,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 11:57:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-11-10 10:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-10 11:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-10 11:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-10 10:12:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-10 11:41:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-11-10 11:57:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05c5a3a9ae921e7eba8ffe9811ffbc3f87f95dfb","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-11-10 05:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e28d45584627751ea13edbccd1160b6a37e3c54b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 05:36:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/31a6e863e8c04ec5ada622c718c32a87 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3954c72023435ca1992936f2393cd8 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00f8e1b1f61b40b1a1069d2070037636 : SUCCESS in 4m 09s\n- openstack-ansible-ssl-galera_server-opensuse-15 https://zuul.opendev.org/t/openstack/build/971b01269ba34828a805ff8f42f09d90 : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95dd1060fe509784dc455145dc8e5ef2731c264f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-10 10:12:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb65c7c665f263cd003e96d1ead797efe2203cc1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-11-10 11:41:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"943d882075e73d911518cb198e5aa8ab2ed7abac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:41:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f965030cf19e89e116ef5814866b9ddd3e481256","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:57:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4952aa3bd67c25c88802fa6f910eaa98532dfa8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:57:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c70df464879747bd8d9da16cf01de949 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7837e8ec2ef3435485f5f9f425cdbda5 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b600765fdb0a450c97cb5445bf8ed4ce : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9b74cb09e417dff7f7086bb82ff9195cd4b29c77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 11:59:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31a3beb943d40e999aa215017b4e71f : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e33a40afeeb74984ad3eaab16a8d8781 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f7b333d95be2378ede9aa8451d9f5068e2c5e06","revisions":{"3f7b333d95be2378ede9aa8451d9f5068e2c5e06":{"kind":"REWORK","_number":1,"created":"2020-11-10 05:11:27.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/41/762041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_client","ref":"refs/changes/41/762041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/41/762041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/41/762041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/41/762041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_client refs/changes/41/762041/1"}}},"commit":{"parents":[{"commit":"f822e7d137cb320550b469779fbb99549f7b552d","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/f822e7d137cb320550b469779fbb99549f7b552d"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-11-10 05:11:07.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-11-10 05:11:24.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nChange-Id: I0d712c73e870e01ed5683bc6f31de5b3100de622\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/3f7b333d95be2378ede9aa8451d9f5068e2c5e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_client/commit/3f7b333d95be2378ede9aa8451d9f5068e2c5e06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
