)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~1004793","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I39e314a687ef75ff622c5ca313d6ffefbdcbacf7","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"hotfix/incr_args","attention_set":{"39442":{"account":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"last_update":"2026-09-22 11:00:12.000000000","reason":"A robot voted negatively on a label"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-24 15:43:35.000000000","reason":"Vote got outdated and was removed: Workflow+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I39e314a687ef75ff622c5ca313d6ffefbdcbacf7","subject":"Fix Mariabackup incremental compression arguments","status":"NEW","created":"2026-09-09 10:02:09.000000000","updated":"2026-09-24 18:11:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab13d5edb2cb11868c1d6e16cfdc05b63f4d6376","_number":1004793,"virtual_id_number":1004793,"owner":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-24 18:11:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":2,"date":"2026-09-24 15:43:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-24 15:43:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-09-24 15:43:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 10:12:07.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-09 12:19:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-11 16:03:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"27375d9eda0bb6542a4bf7268dbb87995407900b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"date":"2026-09-09 10:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b5484afe1a335c21bee3dbdf0235a18cfffcf0c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-09 10:12:07.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"878204be37afc8a4ff640b7a7f12ef0313447e43","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-09 10:12:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d567fe7e78f8de7556c441bf142d398d4ed1326e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:19:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd4f78cbe6834ab1a2d08cacfacc477d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e67bfaf8929469e99f722e290097c83 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1da8753ed8274c33b5043e8e48d64d0a : SUCCESS in 2m 32s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/990a42ea96fb458eabbcc355c08a9807 : FAILURE in 14m 19s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3aceda9fb5b047cd844f6ed32a9b3619 : SUCCESS in 13m 35s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2142f4f0977540d18267d14a911186c5 : SUCCESS in 11m 52s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c769d9dc14954d2b8e959f16fc714ca7 : SUCCESS in 8m 56s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4f274fb0e7bd47eba6fd3e110b0893e9 : SUCCESS in 13m 56s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7213799702624c94a1013b9f88137069 : FAILURE in 6m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/7b5ae2bef652485c9a3b7f01e7587181 : SUCCESS in 1h 30m 51s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/db94bbaf55cc4d2e88f2cb4b73fad618 : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/039e931e132a4706b90a25184eea52ec : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/83d97131c8b1407d91dba21d250570d0 : SUCCESS in 1h 38m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75a293a8478140cba44c4373d6ee9758 : SUCCESS in 1h 45m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4481314bb3284e2788be8de37d05f8a6 : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86929d286bd34b038e01e21e92e87e8c : SUCCESS in 1h 58m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69f12ac98248425caa175910afaa4f6b : SUCCESS in 1h 44m 32s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0d3038bc68ca46aca59f91484e1e5b2a : SUCCESS in 1h 36m 53s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a58c68e77334fbdb1391caf5d6be50a : SUCCESS in 1h 26m 33s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b7059e067bdc44b2b840e09060f0d9f1 : SUCCESS in 1h 04m 09s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6518b73409a415a8bffd691480c65f7 : SUCCESS in 2h 06m 11s","accounts_in_message":[],"_revision_number":1},{"id":"505d7b7930e3eb213e807369db3d14d00a755d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"date":"2026-09-09 15:24:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0cd4e633b9ac5237e661c5b85ba4f345bd19f8ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 17:53:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9316943983345ea9b6a1bd6be0f5439\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fce90ab90154edea227c8ee5ac216c2 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45c4825766b9428eb85cb42821e0ea22 : SUCCESS in 4m 34s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bf2800a50264427b5cf293d7fe91332 : SUCCESS in 14m 45s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/fe9f4079ae6841d485998eeda8b79bd1 : SUCCESS in 12m 24s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/010475fc558847039f02312f20e92274 : SUCCESS in 13m 52s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a962326bf8f5465989392cbee6df51a1 : SUCCESS in 11m 29s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4f5349989de84ddda2cda8ebe197bc86 : SUCCESS in 13m 10s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f9ec886add7494682e49489339b6afc : FAILURE in 8m 03s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a1db10e417c54c96b96bafef4194735d : SUCCESS in 1h 31m 01s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2d97a4dbe46d4c0dac4c0ab24fb06d74 : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/78cd6570a6ce4fd6954f868af7c5fc5f : SUCCESS in 1h 38m 07s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/972e33d30b914d889f38113a1fa7fbb2 : SUCCESS in 1h 07m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e25acd394844df6af33a0d75793d298 : SUCCESS in 1h 41m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f9613a657ad2406e92b8a093f7efa6ce : SUCCESS in 1h 33m 03s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69ffd992d8904fee9599e4d0a8c4a410 : SUCCESS in 1h 53m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36cd1ee198384e45a2b105344a5b2505 : SUCCESS in 1h 34m 00s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c8fe0d9d0db74214a55e59ca17b0b304 : SUCCESS in 1h 31m 37s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bddc1346d6f43b68b0330a100c4c5c1 : SUCCESS in 1h 39m 10s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6dbb23709cdf44eb9f68593136397e32 : SUCCESS in 1h 32m 59s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6650fb1633ad418eaf7f8a62b3b69a1d : SUCCESS in 2h 09m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5028681ddcebacca56513afb274686e41a13922e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-11 16:03:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fd818ca86a2a93bebddedc3be4522b1bbbbb53c9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-21 09:07:59.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9bc187a3777a3bc8d4d52691538f78c7dea917fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 09:12:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c93fb53f56c7709bdde730012bf6306ab2b4dff1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:49:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe7620858cea46ecba8abbfe824a1d52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea08d0c61e8843a7989c0f95bbbda2ab : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2591018b32c648e79603941db4bbcd2d : SUCCESS in 4m 31s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c608edba9d441499eae0fb66a86cd05 : SUCCESS in 15m 18s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0580d9ee27ac4d22a19a21b68e34d8f2 : SUCCESS in 12m 32s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49b9efa3466a428394de49a1b6f60fcd : SUCCESS in 13m 14s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2a3b5e9ec31647b4bd01708d1cd0cf8f : SUCCESS in 13m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/62886f11ef0640eab83518b5f1c5c5c6 : FAILURE in 5m 15s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/45b6791b9b054155934692ecece151fe : SUCCESS in 1h 29m 33s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/85a04c4af4b54d57ba9c0a427f984185 : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6ac778190da44cf58034beb61e3226c0 : SUCCESS in 1h 38m 43s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a8d85afa6a0e4754b002cdce4c517900 : SUCCESS in 1h 36m 58s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38903ff6d0d84545bf0e2fda4eabe1fb : SUCCESS in 1h 29m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fa3336464b57451b9bb3dd7bd205b839 : SUCCESS in 1h 40m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8761840af2a4d7eaea09c27777ff409 : SUCCESS in 1h 59m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a70b7a401ba4c4ab8560e20b690c336 : SUCCESS in 1h 39m 10s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d63b5b5268de4380904e4c30befc8cce : SUCCESS in 1h 33m 59s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a311b7dfb55484187095ee06d795640 : SUCCESS in 1h 36m 34s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fd5175b9422d4d9197d3491f55ad8827 : SUCCESS in 1h 37m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac47fddfb4eb414d915d1789db17bd3e : SUCCESS in 2h 27m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8df79d05fc5650d97ced0f79e2720a2bdc9ff410","author":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"date":"2026-09-22 08:57:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"234b79392517ed4b8a9cfbde5bf66343e51aeced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 11:00:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ad26a8b6f4144c08673ad17d7d7a200\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aefc45e3b5af40c885f3e35057e164b0 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/890c32b261344f7d8ae51817c78ab5a8 : SUCCESS in 2m 42s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9ba0376da2d4f2e8f6448c1d4068067 : SUCCESS in 10m 30s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/53018f7469e046b68f7eee98c189980d : SUCCESS in 16m 08s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6590d6f38994375bcdb77fee72fffb8 : SUCCESS in 11m 31s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e8f70bd72e074143b210e0fc7750c449 : SUCCESS in 13m 14s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a1fc7c1199fd417ca93624492c753d85 : FAILURE in 7m 06s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5eb2a172ed4a48a0823775511b92d10b : FAILURE in 8m 53s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cc07d2e62f3746c39fcd5fcdc1336564 : SUCCESS in 1h 11m 12s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9587bbcc91e549c99823c822b7e7efb3 : SUCCESS in 1h 41m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f6fdbd87cec14ac69ed205f861492317 : SUCCESS in 1h 12m 38s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3fbd3f426f4f4becbbe7a40a7bb160c2 : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f57933a2c924a959516187e95f14066 : SUCCESS in 1h 07m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c676d04a86324b1da2e49c1b910976b7 : SUCCESS in 1h 40m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95a0ecf969a84eb38d89dad67690561a : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb96482d3e2f486f893e9a5023400751 : SUCCESS in 1h 45m 54s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b317bba7be6d4fafbc6b112abfb8b998 : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/031e0de163ba4937b2d86d4f833bb89f : SUCCESS in 1h 02m 18s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1f0725c07f564cebb63162955f530bd7 : SUCCESS in 1h 05m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c4d3836fd9b4e23a0262176df092cbc : SUCCESS in 1h 53m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9d044eb670d5fb607bfb6d6f519e51ae47f1a97b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 11:32:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e2990f99090cc617765d8b33ef672f0627a3af2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 13:48:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b321363cc26d488096eab78e53061feb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb26c915df8d45beb44d26a977d61816 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d84e91ec164cc286703e789b159962 : SUCCESS in 3m 08s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/548669c5bd1d412096e7bf1251f2b11c : SUCCESS in 12m 55s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/e537427256bd475f8f25c1f5bab4a28b : SUCCESS in 11m 23s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53a0f28d8f9e4a3683f939d2aadfdfbe : SUCCESS in 12m 30s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2b8d22722eed43918372dea4224a29fa : SUCCESS in 7m 41s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/e1628502c209410b82c8b3001a96fdc6 : FAILURE in 11m 34s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b2a1acc9afd4411198787de25cfe4876 : FAILURE in 5m 29s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6dde31bf682840a4b1b2d31c9acf7532 : SUCCESS in 1h 39m 58s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a21e1cbcc41e463ebc73a720f4a51faf : SUCCESS in 1h 40m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0b9b3cc97cdc42dfbd392491cad245a8 : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/30e26eba746c498b986478511920bab2 : SUCCESS in 1h 51m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffb24038c8024311ad0d2169a67204b5 : SUCCESS in 1h 31m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/857b639082f64196880c030bd47d8460 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a47153d2fee4891a48db6443423aed4 : SUCCESS in 1h 42m 00s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94fa625a8231494480a18e2a88759eb6 : SUCCESS in 1h 37m 44s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/161891b8784d406681f171880431137a : SUCCESS in 1h 28m 54s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80b61d3beee14a6593426e88d0dd632b : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/646c6ad57f8848b6ad032408011999db : SUCCESS in 1h 06m 14s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc66ba41b1db4c64970d01a76b551b9a : SUCCESS in 2h 07m 42s","accounts_in_message":[],"_revision_number":2},{"id":"7e0de78bdd9570c0b8f825836f5bf1eb63cda687","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 15:43:35.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_39442\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"}],"_revision_number":3},{"id":"ab13d5edb2cb11868c1d6e16cfdc05b63f4d6376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 18:11:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15e795b5edac43a9b86f8618f7dd404b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0c71a13a6642b7a84ed4fc65c9b677 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1099db13f754e92a03233dac5dec346 : SUCCESS in 3m 24s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62940b4250184068b765bbb41b0750c7 : SUCCESS in 16m 12s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/da9382b9bb4f4d4fb293a4f0a077bf0c : SUCCESS in 7m 54s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7170e525de9948ed948a438cf041a692 : SUCCESS in 13m 24s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9cbc2e59de5546489105be562ce8b848 : SUCCESS in 14m 10s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/6a6f54b3db8640d0a65dd0532af81da2 : SUCCESS in 11m 44s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/276d32cf76344be6880f24ea6a49496f : FAILURE in 6m 14s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/89dd9009664944e3881c68b1b4974644 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ae8a4be020c949da9a27c6d9fd46573f : SUCCESS in 1h 32m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4fba44a74bda4202aff0d04a97a11542 : SUCCESS in 1h 30m 23s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f5f2179ed8cb49c69bf127e2498499eb : SUCCESS in 1h 35m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3806f80ce1c4331889648bfa847f3aa : SUCCESS in 1h 25m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/73ec30ad0f9b4b2a99231077b3c0e246 : SUCCESS in 59m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/985122fe55ee4e0f86d0be29bde43251 : FAILURE in 1h 46m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69a8bf835d2449959cf5ff0038122f08 : SUCCESS in 1h 47m 24s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ccd4e1f22e7a4ff89acd0f137a365f2f : SUCCESS in 1h 26m 22s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba6769624f764acba02bb30d939366bf : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/49e1fd8223f74382ad27e0f0329f995c : SUCCESS in 57m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93b1d04f22504e3a97ea602de3b52d35 : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"19cbb62f8aa111e9ac2f716a8c368bd1a7b619be","revisions":{"086054354c6a58bc5744506634c62033a7b264e1":{"kind":"REWORK","_number":1,"created":"2026-09-09 10:02:09.000000000","uploader":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"ref":"refs/changes/93/1004793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/93/1004793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/1"}}},"commit":{"parents":[{"commit":"8f4330aacba730d47c8fa6e74821f95a0caaab37","subject":"Run mariadb-install-db with mysql user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8f4330aacba730d47c8fa6e74821f95a0caaab37"}]}],"author":{"name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","date":"2026-09-09 10:01:16.000000000","tz":120},"committer":{"name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","date":"2026-09-09 10:01:16.000000000","tz":120},"subject":"Fix Mariabackup incremental compression arguments","message":"Fix Mariabackup incremental compression arguments\n\nWe have changed the argument type for compression, meaning\nit no longer expects any true/false value leading to current\nincremental backups failing.\n\nThis change aligns compression handling for incremental backups\nto match the compression handling of full backups.\n\nChange-Id: I39e314a687ef75ff622c5ca313d6ffefbdcbacf7\nSigned-off-by: Christoffer Carlberg \u003cchristoffer.carlberg@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/086054354c6a58bc5744506634c62033a7b264e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/086054354c6a58bc5744506634c62033a7b264e1"}]},"branch":"refs/heads/master"},"8817ceec56542517461dbc2792245a84f34f5cb2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-09 15:24:32.000000000","uploader":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"ref":"refs/changes/93/1004793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/93/1004793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/2"}}},"commit":{"parents":[{"commit":"bdcb03aaaf557468262e6e8e53b6a4c7a5266d8b","subject":"Use ansible.mariadb instead of community.mysql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bdcb03aaaf557468262e6e8e53b6a4c7a5266d8b"}]}],"author":{"name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","date":"2026-09-09 10:01:16.000000000","tz":120},"committer":{"name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","date":"2026-09-09 15:24:32.000000000","tz":0},"subject":"Fix Mariabackup incremental compression arguments","message":"Fix Mariabackup incremental compression arguments\n\nWe have changed the argument type for compression, meaning\nit no longer expects any true/false value leading to current\nincremental backups failing.\n\nThis change aligns compression handling for incremental backups\nto match the compression handling of full backups.\n\nChange-Id: I39e314a687ef75ff622c5ca313d6ffefbdcbacf7\nSigned-off-by: Christoffer Carlberg \u003cchristoffer.carlberg@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8817ceec56542517461dbc2792245a84f34f5cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/8817ceec56542517461dbc2792245a84f34f5cb2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8f4330aacba730d47c8fa6e74821f95a0caaab37","ours":"086054354c6a58bc5744506634c62033a7b264e1","theirs":"bdcb03aaaf557468262e6e8e53b6a4c7a5266d8b","merge_strategy":"recursive","contains_conflicts":false}},"19cbb62f8aa111e9ac2f716a8c368bd1a7b619be":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-24 15:43:35.000000000","uploader":{"_account_id":39442,"name":"Christoffer Carlberg","display_name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","username":"ccarlberg"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/93/1004793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/93/1004793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/93/1004793/3"}}},"commit":{"parents":[{"commit":"2fea14129eaad771cbb40a23f37e1f7e2852a3b8","subject":"Address ansible-galaxy fetching issues.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2fea14129eaad771cbb40a23f37e1f7e2852a3b8"}]}],"author":{"name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","date":"2026-09-09 10:01:16.000000000","tz":120},"committer":{"name":"Christoffer Carlberg","email":"christoffer.carlberg@cleura.com","date":"2026-09-24 15:43:35.000000000","tz":0},"subject":"Fix Mariabackup incremental compression arguments","message":"Fix Mariabackup incremental compression arguments\n\nWe have changed the argument type for compression, meaning\nit no longer expects any true/false value leading to current\nincremental backups failing.\n\nThis change aligns compression handling for incremental backups\nto match the compression handling of full backups.\n\nChange-Id: I39e314a687ef75ff622c5ca313d6ffefbdcbacf7\nSigned-off-by: Christoffer Carlberg \u003cchristoffer.carlberg@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/19cbb62f8aa111e9ac2f716a8c368bd1a7b619be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/19cbb62f8aa111e9ac2f716a8c368bd1a7b619be"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"bdcb03aaaf557468262e6e8e53b6a4c7a5266d8b","ours":"8817ceec56542517461dbc2792245a84f34f5cb2","theirs":"2fea14129eaad771cbb40a23f37e1f7e2852a3b8","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
