)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~1006915","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I80f710978a7cb12cb4421c023666ddedf6912f0f","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"osa/molecule","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-24 09:40:30.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I80f710978a7cb12cb4421c023666ddedf6912f0f","subject":"Address ansible-galaxy fetching issues.","status":"NEW","created":"2026-09-23 09:51:19.000000000","updated":"2026-09-24 11:47:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9497fac39c08b85876db61e2777eb924dceb5e27","_number":1006915,"virtual_id_number":1006915,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 11:47:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-23 13:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"212b30f754e9d6288fa80e0f1f4a214698578a39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 09:51:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"686dcb1f5319556a2a2b977d866f2cfb1b66994c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:00:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9ce4ecb67054cd591b08af8ae9f2ca2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63fdf16ebc04e2086a7a15f796a2429 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32f9efbddd4a48d09aff9f9d671bb32a : SUCCESS in 4m 39s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc64739690f14fc7bdda056e9e7c7c38 : SUCCESS in 20m 14s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c87125932a7c4121a71f5d0ed52be412 : SUCCESS in 12m 24s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/903331039a7943d18cd1b935b65a7857 : SUCCESS in 12m 28s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/360292b6604f488492a190aacdedc1d4 : RETRY_LIMIT in 3m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/67f943dcbe724f428f320344f2d41b46 : SUCCESS in 12m 56s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b678b456f1994b8da602e1f6de5e84fe : FAILURE in 8m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b0c6d37a48fb463aa8c117f5c2739a83 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/46e2a4c6cfcf414bb6c24059b9f5e5c9 : SUCCESS in 1h 42m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/668af96d82ef4014bd089044610f5d18 : SUCCESS in 1h 40m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/48f70f6fcb27458a962954eb50969663 : SUCCESS in 1h 55m 53s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2081cdd2cf644128f00eceace3ea70a : SUCCESS in 1h 41m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/74583d8fde964502b4d91d14fe4b09c6 : SUCCESS in 1h 46m 31s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0542a23d2f4949efa0267e36fb4ee8c2 : SUCCESS in 1h 47m 07s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/010c3aed9a2941c8bdf22fd8fadd982d : SUCCESS in 1h 39m 03s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/458a0867904644119e9423b15398ee96 : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/347ff310a52942c0ae4d0a0a48603e6b : SUCCESS in 1h 45m 40s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/94805c4eb47d4c9b963452c3a9b63126 : SUCCESS in 1h 36m 46s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91bf91c944a848dda98683cd4a43cff7 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"83e3116fc11ca4e8b312bdb86b951a171f4c5d21","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 09:40:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9497fac39c08b85876db61e2777eb924dceb5e27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 11:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b3f22511090488b8a3bd67db1e322fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333a82d64a5c49c9a4644de1fb9acdfa : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/175d61cef7a543c9badd981c46c67f0f : SUCCESS in 4m 50s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c559be1716d4b45a9b310c7badaebe4 : SUCCESS in 19m 27s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/e4c6d92609c24bd19cc03e40edf4b3f5 : SUCCESS in 12m 12s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b23299d2b0e549fba82ba60daa1f554b : SUCCESS in 15m 13s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/942554026f2b4c79a3899e62f682a5bc : SUCCESS in 9m 45s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8aa72c2fb14d4b90b97bf1f1a4133199 : SUCCESS in 13m 04s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/275c1a5e65254a949b5f7ac1fe9ebc9a : FAILURE in 5m 53s (non-voting)\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d575373f5f14b47be867a7c2ff4ada5 : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3b848091f2924f39b3c449472b00246e : SUCCESS in 1h 33m 48s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b9d54b71386d4498add82ac059edd416 : SUCCESS in 1h 47m 13s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1e737391dc0c4a70a69d2e8eadee911d : SUCCESS in 1h 31m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aaa4d479efc49e9b6b9b9164dbc12fe : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f6435fd06c9a403fa689fe79646cac55 : SUCCESS in 1h 42m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9991ea02abf44536a040cacb783418f9 : SUCCESS in 1h 47m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd60686fa9e44d998f02e115905c8f62 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-infra_distro_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/ec13f2aff24d44129d31c0260bcbf863 : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebbf9f385d6d493ebdf8a7015b73787e : SUCCESS in 1h 37m 47s\n- openstack-ansible-deploy-infra_distro_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/138d232ed54e406ba65a09919dcb4393 : SUCCESS in 1h 30m 10s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5f85b960e874feda72514c39ce1502d : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fea14129eaad771cbb40a23f37e1f7e2852a3b8","revisions":{"2fea14129eaad771cbb40a23f37e1f7e2852a3b8":{"kind":"REWORK","_number":1,"created":"2026-09-23 09:51:19.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/1006915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/15/1006915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/1006915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/1006915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/1006915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/15/1006915/1"}}},"commit":{"parents":[{"commit":"bdcb03aaaf557468262e6e8e53b6a4c7a5266d8b","subject":"Use ansible.mariadb instead of community.mysql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bdcb03aaaf557468262e6e8e53b6a4c7a5266d8b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 09:50:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-23 09:50:34.000000000","tz":120},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with\nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I80f710978a7cb12cb4421c023666ddedf6912f0f\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2fea14129eaad771cbb40a23f37e1f7e2852a3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/2fea14129eaad771cbb40a23f37e1f7e2852a3b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
