)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~576780","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6a533883175d54daf6dc03c477e35100e097c17a","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I6a533883175d54daf6dc03c477e35100e097c17a","subject":"Trivial: Fix the pep8 warning","status":"MERGED","created":"2018-06-20 08:08:44.000000000","updated":"2019-04-24 19:33:12.000000000","submitted":"2019-04-24 19:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"576780-1556134392778-0aef32fb","meta_rev_id":"72b634e1328d1ca7a8f94eeb6617eb497219ae3c","_number":576780,"virtual_id_number":576780,"owner":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 19:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},{"value":2,"date":"2019-04-24 18:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},{"value":1,"date":"2019-04-24 18:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-11 07:33:23.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-24 18:56:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 19:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5540e723ebd4e810e596f5f87d4a91b2c116acd","author":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"date":"2018-06-20 08:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c03352d8b2da38c1739887f822b0e393faa9383","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-20 09:44:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/80/576780/1/check/build-openstack-sphinx-docs/749e863/html/ : SUCCESS in 4m 59s\n- openstack-ansible-linters http://logs.openstack.org/80/576780/1/check/openstack-ansible-linters/6110f06/ : SUCCESS in 8m 11s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-centos-7/4e2bc2b/ : TIMED_OUT in 1h 32m 47s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-150/93d40c1/ : FAILURE in 10m 35s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-423/24311d3/ : SUCCESS in 19m 08s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-bionic/bb9dcaf/ : SUCCESS in 23m 49s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-xenial/de86b92/ : SUCCESS in 18m 34s\n- openstack-ansible-upgrade-opensuse-150 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-opensuse-150/ffeefe4/ : FAILURE in 9m 18s (non-voting)\n- openstack-ansible-upgrade-opensuse-423 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-opensuse-423/303c7b5/ : SUCCESS in 43m 20s\n- openstack-ansible-upgrade-ubuntu-bionic http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-ubuntu-bionic/146ec81/ : FAILURE in 15m 53s (non-voting)\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-ubuntu-xenial/4a6aeae/ : SUCCESS in 31m 11s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-centos-7-ssl/60e57f6/ : SUCCESS in 23m 38s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-423-ssl/ef7fc63/ : SUCCESS in 20m 21s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/b92c8e8/ : SUCCESS in 21m 00s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/50ffaf0/ : SUCCESS in 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3caca890b77ef6506d880a29c97e6fd83e50bd11","author":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"date":"2018-06-20 10:05:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"935635205a6d39c87dd1d752f23cf9b2577864f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-20 10:42:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/80/576780/1/check/build-openstack-sphinx-docs/0194ae0/html/ : SUCCESS in 2m 58s\n- openstack-ansible-linters http://logs.openstack.org/80/576780/1/check/openstack-ansible-linters/3d9beb3/ : SUCCESS in 5m 59s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-centos-7/364a785/ : SUCCESS in 23m 02s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-150/51661fb/ : FAILURE in 8m 53s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-423/ef6fc5e/ : SUCCESS in 17m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-bionic/4262e45/ : SUCCESS in 22m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-xenial/70c2dbf/ : SUCCESS in 17m 25s\n- openstack-ansible-upgrade-opensuse-150 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-opensuse-150/0a93d5a/ : FAILURE in 11m 38s (non-voting)\n- openstack-ansible-upgrade-opensuse-423 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-opensuse-423/3ddfc37/ : SUCCESS in 31m 24s\n- openstack-ansible-upgrade-ubuntu-bionic http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-ubuntu-bionic/c73b85e/ : FAILURE in 17m 14s (non-voting)\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-ubuntu-xenial/1882a70/ : SUCCESS in 26m 11s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-centos-7-ssl/76449e1/ : SUCCESS in 26m 41s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-423-ssl/e19456c/ : SUCCESS in 20m 05s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/386cda2/ : SUCCESS in 18m 21s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/1ae5141/ : SUCCESS in 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4c9f83159c7c0e96160bbf54f4520a2c05d15723","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2018-06-26 13:06:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"60e215d9947d53b426273524cc848757784f1510","author":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"date":"2018-07-10 09:19:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de9e2db96939eadbbf7c68b212598405687bc512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 09:49:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/80/576780/1/check/build-openstack-sphinx-docs/945b182/html/ : SUCCESS in 3m 40s\n- openstack-ansible-linters http://logs.openstack.org/80/576780/1/check/openstack-ansible-linters/c4d25cb/ : SUCCESS in 6m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-centos-7/125e863/ : SUCCESS in 23m 43s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-150/c4ff64d/ : FAILURE in 16m 28s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-423/e6a2839/ : SUCCESS in 19m 42s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-bionic/7b87ace/ : SUCCESS in 21m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-xenial/b13922e/ : SUCCESS in 21m 40s\n- openstack-ansible-upgrade-centos-7 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-centos-7/e838049/ : FAILURE in 18m 29s (non-voting)\n- openstack-ansible-upgrade-opensuse-150 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-opensuse-150/676b7fd/ : FAILURE in 17m 18s (non-voting)\n- openstack-ansible-upgrade-opensuse-423 http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-opensuse-423/487b3f0/ : FAILURE in 15m 07s (non-voting)\n- openstack-ansible-upgrade-ubuntu-bionic http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-ubuntu-bionic/eab8b98/ : FAILURE in 16m 43s (non-voting)\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/80/576780/1/check/openstack-ansible-upgrade-ubuntu-xenial/fb19a8a/ : FAILURE in 15m 51s (non-voting)\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-centos-7-ssl/1883e96/ : SUCCESS in 23m 09s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-opensuse-423-ssl/b4342cb/ : SUCCESS in 21m 18s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/f0facf7/ : SUCCESS in 20m 42s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/80/576780/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/ecf177b/ : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"68096a5645c348de6aa1da248e772e75915588a7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-07-11 07:33:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b5d033c7d38058055a2cccedb3411da6a6e96d9","author":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"date":"2018-07-11 07:35:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1711a74c30c8cd9bccce9264b500e6677913425b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-11 08:11:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/80/576780/2/check/build-openstack-sphinx-docs/36ad2b2/html/ : SUCCESS in 3m 08s\n- openstack-ansible-linters http://logs.openstack.org/80/576780/2/check/openstack-ansible-linters/881471c/ : SUCCESS in 8m 09s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-centos-7/196b7fa/ : SUCCESS in 23m 05s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-opensuse-150/e853fd2/ : FAILURE in 17m 32s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-opensuse-423/3d24052/ : SUCCESS in 20m 52s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-ubuntu-bionic/72bda24/ : SUCCESS in 26m 30s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-ubuntu-xenial/7d55c88/ : SUCCESS in 19m 55s\n- openstack-ansible-upgrade-centos-7 http://logs.openstack.org/80/576780/2/check/openstack-ansible-upgrade-centos-7/10d590c/ : FAILURE in 17m 44s (non-voting)\n- openstack-ansible-upgrade-opensuse-150 http://logs.openstack.org/80/576780/2/check/openstack-ansible-upgrade-opensuse-150/b165d8d/ : FAILURE in 18m 13s (non-voting)\n- openstack-ansible-upgrade-opensuse-423 http://logs.openstack.org/80/576780/2/check/openstack-ansible-upgrade-opensuse-423/b2c434a/ : FAILURE in 16m 01s (non-voting)\n- openstack-ansible-upgrade-ubuntu-bionic http://logs.openstack.org/80/576780/2/check/openstack-ansible-upgrade-ubuntu-bionic/64f7944/ : FAILURE in 16m 42s (non-voting)\n- openstack-ansible-upgrade-ubuntu-xenial http://logs.openstack.org/80/576780/2/check/openstack-ansible-upgrade-ubuntu-xenial/ea01665/ : FAILURE in 14m 16s (non-voting)\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-centos-7-ssl/10ee16f/ : SUCCESS in 25m 40s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-opensuse-423-ssl/3ad656e/ : SUCCESS in 19m 03s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-ubuntu-xenial-ssl/569ae3c/ : SUCCESS in 18m 41s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/80/576780/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/6349f91/ : SUCCESS in 25m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7810409498ee77c71f7cc0c44ec73b19b5ad69bc","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2018-07-11 14:46:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"21b65568e9dc952e38932f7817bf07b38cfa4ce6","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-01-08 13:51:59.000000000","message":"Removed reviewer kourosh vivan with the following votes:\n\n* Code-Review+1 by kourosh vivan \u003ckourosh.vivan@osones.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"ae877215eff91f6d48118b04a60fb6c69433794b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:56:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4135b068a838e1e038a19d90f6511642d166ff4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:01:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4f78a309e6ba09b4355ef8ef0fc879b45ebd4d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:33:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/80/576780/2/gate/openstack-ansible-linters/135c70c/ : SUCCESS in 6m 15s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/80/576780/2/gate/openstack-ansible-functional-debian-stable/53cf80d/ : SUCCESS in 14m 39s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/576780/2/gate/openstack-ansible-functional-centos-7/3554fa5/ : SUCCESS in 18m 09s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/80/576780/2/gate/openstack-ansible-functional-ubuntu-bionic/7a9c9d6/ : SUCCESS in 17m 50s\n- openstack-tox-docs http://logs.openstack.org/80/576780/2/gate/openstack-tox-docs/ea57991/html/ : SUCCESS in 2m 55s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/80/576780/2/gate/openstack-ansible-functional-centos-7-ssl/9c840bc/ : SUCCESS in 18m 44s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/80/576780/2/gate/openstack-ansible-functional-ubuntu-xenial-ssl/7e444af/ : SUCCESS in 17m 05s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/80/576780/2/gate/openstack-ansible-functional-ubuntu-bionic-ssl/dce92df/ : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":2},{"id":"72b634e1328d1ca7a8f94eeb6617eb497219ae3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:33:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0820e751964dad1b870f8610575e20273a35688","revisions":{"fbc7fb45a4a4462a72cf75370ce54ed281a02bc6":{"kind":"REWORK","_number":1,"created":"2018-06-20 08:08:44.000000000","uploader":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"ref":"refs/changes/80/576780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/80/576780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/1"}}},"commit":{"parents":[{"commit":"c5be0195e8708ef1ed9fb84135aa2c8cb65eefa8","subject":"Merge \"zuul: Add openSUSE upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c5be0195e8708ef1ed9fb84135aa2c8cb65eefa8"}]}],"author":{"name":"fpxie","email":"fpxie@fiberhome.com","date":"2018-06-20 08:06:44.000000000","tz":480},"committer":{"name":"fpxie","email":"fpxie@fiberhome.com","date":"2018-06-20 08:08:16.000000000","tz":480},"subject":"Trivial: Fix the pep8 warning","message":"Trivial: Fix the pep8 warning\n\nThe yaml should start with \"---\"\n\nChange-Id: I6a533883175d54daf6dc03c477e35100e097c17a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fbc7fb45a4a4462a72cf75370ce54ed281a02bc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/fbc7fb45a4a4462a72cf75370ce54ed281a02bc6"}]},"branch":"refs/heads/master"},"a0820e751964dad1b870f8610575e20273a35688":{"kind":"REWORK","_number":2,"created":"2018-07-11 07:35:18.000000000","uploader":{"_account_id":25284,"name":"fupingxie","email":"xiefp88@sina.com","username":"fpxie"},"ref":"refs/changes/80/576780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/80/576780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/80/576780/2"}}},"commit":{"parents":[{"commit":"c5be0195e8708ef1ed9fb84135aa2c8cb65eefa8","subject":"Merge \"zuul: Add openSUSE upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c5be0195e8708ef1ed9fb84135aa2c8cb65eefa8"}]}],"author":{"name":"fpxie","email":"fpxie@fiberhome.com","date":"2018-06-20 08:06:44.000000000","tz":480},"committer":{"name":"fupingxie","email":"xiefp88@sina.com","date":"2018-07-11 07:34:49.000000000","tz":0},"subject":"Trivial: Fix the pep8 warning","message":"Trivial: Fix the pep8 warning\n\nThe yaml should start with \"---\"\n\nChange-Id: I6a533883175d54daf6dc03c477e35100e097c17a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a0820e751964dad1b870f8610575e20273a35688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/a0820e751964dad1b870f8610575e20273a35688"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
