)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~616530","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I0c6de41fd059f104f060d4634c207c834f28f3dd","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"pep8","hashtags":[],"change_id":"I0c6de41fd059f104f060d4634c207c834f28f3dd","subject":"Trivial: Fix the pep8 warning","status":"MERGED","created":"2018-11-08 12:05:29.000000000","updated":"2019-04-24 19:41:07.000000000","submitted":"2019-04-24 19:41:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"616530-1556134867113-332a2ac9","meta_rev_id":"375b8924e54caa5166047042164407d1bb40b5a8","_number":616530,"virtual_id_number":616530,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 19:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-04-24 18:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-04-24 18:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 18:57:17.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 19:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23c2ef98321aa6579bcfb4d8a706b051eaf9016c","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-11-08 12:05:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e504d9c695cac04aaa3e1846ab2570e41abf6915","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 12:42:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/30/616530/1/check/openstack-ansible-linters/b07dd8f/ : SUCCESS in 6m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-centos-7/d45c86f/ : SUCCESS in 26m 54s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-opensuse-150/8d2be47/ : SUCCESS in 20m 35s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-opensuse-423/0292ab7/ : SUCCESS in 28m 54s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-ubuntu-bionic/b69d0bf/ : SUCCESS in 19m 11s\n- openstack-tox-docs http://logs.openstack.org/30/616530/1/check/openstack-tox-docs/fe3e652/html/ : SUCCESS in 2m 50s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-centos-7-ssl/de3839f/ : SUCCESS in 19m 23s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-opensuse-423-ssl/a088204/ : SUCCESS in 16m 43s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/0b4a79a/ : SUCCESS in 35m 05s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/30/616530/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/388d245/ : SUCCESS in 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f4f5d0909085dd2bf4353b93627da6f8f451985a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:57:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f30d82d715f89881fe5cc392c05578b1bfd9cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:01:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14f7f6712e02fdcc095be28c38ff02572d2e45c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:41:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/30/616530/1/gate/openstack-ansible-linters/d10c0ba/ : SUCCESS in 10m 17s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/30/616530/1/gate/openstack-ansible-functional-debian-stable/0e21583/ : SUCCESS in 28m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/30/616530/1/gate/openstack-ansible-functional-centos-7/4c5caf2/ : SUCCESS in 18m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/30/616530/1/gate/openstack-ansible-functional-ubuntu-bionic/7c5ec06/ : SUCCESS in 18m 13s\n- openstack-tox-docs http://logs.openstack.org/30/616530/1/gate/openstack-tox-docs/3c3a357/html/ : SUCCESS in 5m 28s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/30/616530/1/gate/openstack-ansible-functional-centos-7-ssl/1bdc5f7/ : SUCCESS in 33m 01s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/30/616530/1/gate/openstack-ansible-functional-ubuntu-xenial-ssl/689b288/ : SUCCESS in 29m 09s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/30/616530/1/gate/openstack-ansible-functional-ubuntu-bionic-ssl/9bdeffd/ : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a0a349e206bc6c1993abef790239c52a5052c319","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:41:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b380ad756fcc43d46dca79244a493a6f27a88831","revisions":{"b380ad756fcc43d46dca79244a493a6f27a88831":{"kind":"REWORK","_number":1,"created":"2018-11-08 12:05:29.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/30/616530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/30/616530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/30/616530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/30/616530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/30/616530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/30/616530/1"}}},"commit":{"parents":[{"commit":"ea327ee579d4e4a0bac7bbd2ceee4b55689b8dc5","subject":"Merge \"Fix Galera self-signed SSL functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ea327ee579d4e4a0bac7bbd2ceee4b55689b8dc5"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2018-11-08 12:05:09.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2018-11-08 12:05:27.000000000","tz":480},"subject":"Trivial: Fix the pep8 warning","message":"Trivial: Fix the pep8 warning\n\nThe yaml should start with \"---\"\n\nChange-Id: I0c6de41fd059f104f060d4634c207c834f28f3dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b380ad756fcc43d46dca79244a493a6f27a88831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b380ad756fcc43d46dca79244a493a6f27a88831"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
