)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~654338","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ifad0cdd6c5ea008c056d737dbb1221b32448fb3f","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ifad0cdd6c5ea008c056d737dbb1221b32448fb3f","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:12:23.000000000","updated":"2019-04-22 11:43:41.000000000","submitted":"2019-04-22 11:43:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654338-1555933421668-dfe5b5ea","meta_rev_id":"ab5e6a1a5108bcee942853014d1c2c5c31b9e5bd","_number":654338,"virtual_id_number":654338,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 11:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 11:04:21.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 11:04:21.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 11:04:21.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 11:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cee498f2485ad8e6cae39dab54d7f4d496ac9669","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5252f581266616c867fc3c5b45beee3f1fb4cfc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:39:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/38/654338/1/check/openstack-ansible-linters/608bfba/ : SUCCESS in 6m 01s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/38/654338/1/check/openstack-ansible-functional-debian-stable/b0184ec/ : SUCCESS in 15m 06s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/38/654338/1/check/openstack-ansible-functional-centos-7/897a144/ : SUCCESS in 20m 00s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/38/654338/1/check/openstack-ansible-functional-ubuntu-bionic/ae82faa/ : SUCCESS in 14m 17s\n- openstack-tox-docs http://logs.openstack.org/38/654338/1/check/openstack-tox-docs/42d2cf5/html/ : SUCCESS in 3m 03s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/38/654338/1/check/openstack-ansible-functional-centos-7-ssl/4ab631e/ : SUCCESS in 19m 36s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/38/654338/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/f8038b8/ : SUCCESS in 16m 02s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/38/654338/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/643e248/ : SUCCESS in 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f6b721350bf7eb37ec032f08f062c8b9c4c2bebf","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 11:04:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6959f11598c119d3e36e29558ec5d79f5a42e0a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:04:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90e4a8c71095d31fcdddf571a58dbd076a75b452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:43:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1acbfc033a3441703db6dda8ab19ba5404d3edf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 11:43:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/38/654338/1/gate/openstack-ansible-linters/73e7ef5/ : SUCCESS in 5m 50s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/38/654338/1/gate/openstack-ansible-functional-debian-stable/dd4edf0/ : SUCCESS in 15m 55s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/38/654338/1/gate/openstack-ansible-functional-centos-7/eff190d/ : SUCCESS in 18m 08s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/38/654338/1/gate/openstack-ansible-functional-ubuntu-bionic/8ebdfd1/ : SUCCESS in 16m 17s\n- openstack-tox-docs http://logs.openstack.org/38/654338/1/gate/openstack-tox-docs/fccdaad/html/ : SUCCESS in 3m 02s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/38/654338/1/gate/openstack-ansible-functional-centos-7-ssl/98d1d18/ : SUCCESS in 19m 31s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/38/654338/1/gate/openstack-ansible-functional-ubuntu-xenial-ssl/5de5beb/ : SUCCESS in 15m 25s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/38/654338/1/gate/openstack-ansible-functional-ubuntu-bionic-ssl/ca82104/ : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c501d464481fdeb299e54b15b31f20e8975e1bc","revisions":{"6c501d464481fdeb299e54b15b31f20e8975e1bc":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:12:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/654338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/38/654338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/654338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/654338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/654338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/654338/1"}}},"commit":{"parents":[{"commit":"ada6845a7a786516c130b1d41c5b0abc0b41a69d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ada6845a7a786516c130b1d41c5b0abc0b41a69d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:12:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:12:22.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ifad0cdd6c5ea008c056d737dbb1221b32448fb3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6c501d464481fdeb299e54b15b31f20e8975e1bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6c501d464481fdeb299e54b15b31f20e8975e1bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
