)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~657234","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ia9885316bfea7f97d254042356daba7e328fdaa3","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Ia9885316bfea7f97d254042356daba7e328fdaa3","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-05 11:40:42.000000000","updated":"2019-07-16 17:32:48.000000000","submitted":"2019-07-16 17:32:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657234-1563298368209-0a1ee91c","meta_rev_id":"3b6042caa24b49aa79ef219b933cfc965b6573a8","_number":657234,"virtual_id_number":657234,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-16 17:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"all":[{"value":1,"date":"2019-07-10 15:19:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":2,"date":"2019-07-16 14:26:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-16 15:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 15:19:48.000000000","updated_by":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"reviewer":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"state":"REVIEWER"},{"updated":"2019-07-16 14:26:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-16 15:57:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-16 17:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02e71607fda34bef767a235ba982ce100546a357","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-05 11:40:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9b2373b6b6820b2a59d241416bcb0178c028481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 12:05:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/657234/1/check/openstack-ansible-linters/a4806ff/ : SUCCESS in 4m 30s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/34/657234/1/check/openstack-ansible-functional-debian-stable/093ef13/ : SUCCESS in 16m 32s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/34/657234/1/check/openstack-ansible-functional-centos-7/60393b0/ : SUCCESS in 18m 44s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/34/657234/1/check/openstack-ansible-functional-ubuntu-bionic/2415985/ : SUCCESS in 15m 24s\n- openstack-tox-docs http://logs.openstack.org/34/657234/1/check/openstack-tox-docs/94c397a/html/ : SUCCESS in 2m 39s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/34/657234/1/check/openstack-ansible-functional-centos-7-ssl/5432c3b/ : SUCCESS in 18m 37s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/34/657234/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/3ebe6b7/ : SUCCESS in 15m 39s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/34/657234/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/b9d95ff/ : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b2e94c6778bf0ec4229bf7bcb0aa97692b605dbf","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-06 10:33:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ea7f3aef8542d7a5981aef3e8f63d7a443167bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 11:13:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/657234/2/check/openstack-ansible-linters/2d4cedf/ : SUCCESS in 4m 19s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/34/657234/2/check/openstack-ansible-functional-debian-stable/e39e660/ : SUCCESS in 15m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/34/657234/2/check/openstack-ansible-functional-centos-7/d7240d1/ : SUCCESS in 23m 36s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/34/657234/2/check/openstack-ansible-functional-ubuntu-bionic/450d0e8/ : SUCCESS in 15m 10s\n- openstack-tox-docs http://logs.openstack.org/34/657234/2/check/openstack-tox-docs/2a6f7df/html/ : SUCCESS in 2m 19s\n- build-openstack-releasenotes http://logs.openstack.org/34/657234/2/check/build-openstack-releasenotes/35f2807/html/ : SUCCESS in 2m 15s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/34/657234/2/check/openstack-ansible-functional-centos-7-ssl/e59606e/ : SUCCESS in 18m 50s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/34/657234/2/check/openstack-ansible-functional-ubuntu-xenial-ssl/158a9ad/ : SUCCESS in 17m 57s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/34/657234/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/dd7765e/ : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":2},{"id":"85733725b71b91243bd62871fa665f43a9960932","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-07-10 13:36:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff369c16b9d4f35bbe475f357db5a6e3f1654bfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:07:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/657234/3/check/openstack-ansible-linters/63c517c/ : SUCCESS in 6m 14s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/34/657234/3/check/openstack-ansible-functional-debian-stable/862a3a2/ : SUCCESS in 19m 19s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/34/657234/3/check/openstack-ansible-functional-centos-7/f9de9f5/ : SUCCESS in 37m 01s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/34/657234/3/check/openstack-ansible-functional-ubuntu-bionic/f2b8fa2/ : SUCCESS in 16m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/34/657234/3/check/openstack-ansible-deploy-aio_metal-debian-stable/412a857/ : SUCCESS in 1h 01m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/34/657234/3/check/openstack-ansible-deploy-aio_metal-centos-7/a68984c/ : SUCCESS in 1h 23m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/34/657234/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/3e92060/ : SUCCESS in 1h 24m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/34/657234/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/290161f/ : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/34/657234/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/395c7aa/ : SUCCESS in 1h 00m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/34/657234/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/1526450/ : FAILURE in 25m 42s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/34/657234/3/check/openstack-tox-docs/85edcbd/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/34/657234/3/check/build-openstack-releasenotes/067f308/html/ : SUCCESS in 3m 19s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/34/657234/3/check/openstack-ansible-functional-centos-7-ssl/b25cf5f/ : SUCCESS in 22m 52s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/34/657234/3/check/openstack-ansible-functional-opensuse-15-ssl/dc02fa1/ : FAILURE in 36m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/34/657234/3/check/openstack-ansible-functional-ubuntu-bionic-ssl/07a1cd7/ : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d27ee2c6612298b936c3e8c63492e77fcc40b2b4","author":{"_account_id":18279,"name":"Jonathan Herlin","email":"jonte@jherlin.se","username":"Jonher937"},"date":"2019-07-10 15:19:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0b2826f19142410dafc3b9cfc779bd06b8f4d784","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 14:26:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"25e7fed14653a1981b3769b980e3dfa5abeba58f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-16 15:57:47.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fa676287930b330abcf4bf53761305eb253f143f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:57:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3f073ad92762d440545243046a62300338aaab15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:32:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/657234/3/gate/openstack-ansible-linters/6088c2e/ : SUCCESS in 6m 29s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/34/657234/3/gate/openstack-ansible-functional-debian-stable/18cdf9f/ : SUCCESS in 14m 44s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/34/657234/3/gate/openstack-ansible-functional-centos-7/700a735/ : SUCCESS in 18m 43s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/34/657234/3/gate/openstack-ansible-functional-ubuntu-bionic/8fd05af/ : SUCCESS in 14m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/34/657234/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/ef8e505/ : SUCCESS in 1h 04m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/34/657234/3/gate/openstack-ansible-deploy-aio_metal-centos-7/f27eb7e/ : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/34/657234/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/57c9e2d/ : SUCCESS in 1h 25m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/34/657234/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9f41a3b/ : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/34/657234/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/0640d9b/ : SUCCESS in 1h 08m 54s\n- openstack-tox-docs http://logs.openstack.org/34/657234/3/gate/openstack-tox-docs/1b6c095/html/ : SUCCESS in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/34/657234/3/gate/build-openstack-releasenotes/075d81a/html/ : SUCCESS in 3m 43s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/34/657234/3/gate/openstack-ansible-functional-centos-7-ssl/b93609f/ : SUCCESS in 19m 54s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/34/657234/3/gate/openstack-ansible-functional-ubuntu-bionic-ssl/0b0614e/ : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":3},{"id":"3b6042caa24b49aa79ef219b933cfc965b6573a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:32:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3cc0d9d2958d96a8360bbd746c65e8d7215fbef","revisions":{"e0511bd7c909c6d6c4c06c0441c6538551c071ea":{"kind":"REWORK","_number":1,"created":"2019-05-05 11:40:42.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/34/657234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/34/657234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/1"}}},"commit":{"parents":[{"commit":"c3369aa7178884ef01144a722cd6961c1673e8bd","subject":"Merge \"Trivial: Fix the pep8 warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c3369aa7178884ef01144a722cd6961c1673e8bd"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 11:21:14.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 11:40:41.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia9885316bfea7f97d254042356daba7e328fdaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e0511bd7c909c6d6c4c06c0441c6538551c071ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e0511bd7c909c6d6c4c06c0441c6538551c071ea"}]},"branch":"refs/heads/master"},"9469523b597da55c0420162a79a0f7aedb4f046e":{"kind":"REWORK","_number":2,"created":"2019-05-06 10:33:17.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/34/657234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/34/657234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/2"}}},"commit":{"parents":[{"commit":"c3369aa7178884ef01144a722cd6961c1673e8bd","subject":"Merge \"Trivial: Fix the pep8 warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c3369aa7178884ef01144a722cd6961c1673e8bd"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 11:21:14.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-06 10:33:06.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia9885316bfea7f97d254042356daba7e328fdaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9469523b597da55c0420162a79a0f7aedb4f046e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9469523b597da55c0420162a79a0f7aedb4f046e"}]},"branch":"refs/heads/master"},"e3cc0d9d2958d96a8360bbd746c65e8d7215fbef":{"kind":"REWORK","_number":3,"created":"2019-07-10 13:36:47.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/34/657234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/34/657234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/34/657234/3"}}},"commit":{"parents":[{"commit":"c3369aa7178884ef01144a722cd6961c1673e8bd","subject":"Merge \"Trivial: Fix the pep8 warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c3369aa7178884ef01144a722cd6961c1673e8bd"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-05 11:21:14.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-07-10 13:36:17.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia9885316bfea7f97d254042356daba7e328fdaa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e3cc0d9d2958d96a8360bbd746c65e8d7215fbef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e3cc0d9d2958d96a8360bbd746c65e8d7215fbef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
