)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~659003","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I494d754ae23071e56502d3e99fa5ba603a09e5e3","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"I494d754ae23071e56502d3e99fa5ba603a09e5e3","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-14 09:45:14.000000000","updated":"2019-05-14 22:46:39.000000000","submitted":"2019-05-14 22:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659003-1557873999666-0de1d12d","meta_rev_id":"35771a73474424f92f9cb9e28a5fbadf8137b94d","_number":659003,"virtual_id_number":659003,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-14 22:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-05-14 09:59:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},{"value":1,"date":"2019-05-14 15:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-05-14 09:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 21:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-14 21:17:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 09:56:39.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-14 09:59:28.000000000","updated_by":{"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},"reviewer":{"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-05-14 15:06:05.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-05-14 21:17:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-14 22:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4c662b2184d6a3743ae7038d4e89ebe8804d9b8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-14 09:45:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df7d03186c55a400da6f5f62ef9e957e4f78706f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-14 09:56:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72b5e9c3d9a2030d15bc44051970832ce8a55cc8","author":{"_account_id":30403,"name":"hu","email":"huxy@awcloud.com"},"date":"2019-05-14 09:59:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bede89d314252c3bd057d8f24b0f635ab5908030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 11:16:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/659003/1/check/openstack-ansible-linters/973d2a8/ : SUCCESS in 4m 32s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/03/659003/1/check/openstack-ansible-functional-debian-stable/ea745bd/ : SUCCESS in 16m 10s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/659003/1/check/openstack-ansible-functional-centos-7/c341c1b/ : SUCCESS in 19m 01s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/659003/1/check/openstack-ansible-functional-ubuntu-bionic/0b50308/ : SUCCESS in 16m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/659003/1/check/openstack-ansible-deploy-aio_metal-debian-stable/5d3ccaa/ : SUCCESS in 1h 04m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/659003/1/check/openstack-ansible-deploy-aio_metal-centos-7/78c4e8f/ : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/03/659003/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/dc361b1/ : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/659003/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c39e478/ : SUCCESS in 46m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/03/659003/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8edde9f/ : FAILURE in 38m 15s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/03/659003/1/check/openstack-tox-docs/8b3fc9b/html/ : SUCCESS in 2m 07s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/03/659003/1/check/openstack-ansible-functional-centos-7-ssl/35beb42/ : SUCCESS in 18m 08s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/03/659003/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/ca93cee/ : SUCCESS in 14m 54s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/03/659003/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/4bcc37f/ : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2b465b7c66775e231f890d29a0e7d0b0e3efddb2","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-05-14 15:06:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"59436ff4e1551cdfa5c17769f98c316373c2bbe0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-14 21:17:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e2d0994bc34e337b8f202d51ba61f6817ad3e5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 21:19:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a698b926868d95aad269a9029f577e87acd299b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:46:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/659003/1/gate/openstack-ansible-linters/29c8ee3/ : SUCCESS in 8m 29s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/03/659003/1/gate/openstack-ansible-functional-debian-stable/330af33/ : SUCCESS in 19m 29s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/659003/1/gate/openstack-ansible-functional-centos-7/737f28f/ : SUCCESS in 22m 39s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/659003/1/gate/openstack-ansible-functional-ubuntu-bionic/4fbf1af/ : SUCCESS in 19m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/659003/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/01a6a70/ : SUCCESS in 54m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/659003/1/gate/openstack-ansible-deploy-aio_metal-centos-7/9c670de/ : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/03/659003/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/45cb6e9/ : SUCCESS in 57m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/659003/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/eab43c3/ : SUCCESS in 55m 42s\n- openstack-tox-docs http://logs.openstack.org/03/659003/1/gate/openstack-tox-docs/44f6cf0/html/ : SUCCESS in 2m 52s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/03/659003/1/gate/openstack-ansible-functional-centos-7-ssl/b2ce65d/ : SUCCESS in 21m 17s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/03/659003/1/gate/openstack-ansible-functional-ubuntu-xenial-ssl/2903ab2/ : SUCCESS in 15m 42s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/03/659003/1/gate/openstack-ansible-functional-ubuntu-bionic-ssl/3d426b1/ : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bf32aa416d886e4022e993f9c243c4e7dd3b7828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 22:46:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc7a32976eea219381a3e7b8a140f45dfd8f285b","revisions":{"dc7a32976eea219381a3e7b8a140f45dfd8f285b":{"kind":"REWORK","_number":1,"created":"2019-05-14 09:45:14.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/03/659003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/659003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/659003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/659003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/659003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/659003/1"}}},"commit":{"parents":[{"commit":"3e596051c231e0e0d820e7c6dc95e652056a7e93","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/3e596051c231e0e0d820e7c6dc95e652056a7e93"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 09:43:45.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-14 09:43:45.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I494d754ae23071e56502d3e99fa5ba603a09e5e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dc7a32976eea219381a3e7b8a140f45dfd8f285b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dc7a32976eea219381a3e7b8a140f45dfd8f285b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
