)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~660141","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fqueens~I9fb16b7f046530efc124e760bc7870c13b4e8d32","project":"openstack/openstack-ansible-galera_server","branch":"stable/queens","topic":"percona-release","hashtags":[],"change_id":"I9fb16b7f046530efc124e760bc7870c13b4e8d32","subject":"Update percona-release package","status":"MERGED","created":"2019-05-20 14:52:45.000000000","updated":"2019-05-21 14:27:42.000000000","submitted":"2019-05-21 14:27:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660141-1558448862679-795c03e7","meta_rev_id":"30749ca1021484e5f1f21d37c5e374f5986c4969","_number":660141,"virtual_id_number":660141,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-21 14:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"recommended":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"all":[{"value":1,"date":"2019-05-20 21:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 11:43:15.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-05-20 21:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"date":"2019-05-21 13:51:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 11:43:15.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-05-21 12:22:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 21:50:17.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"},{"updated":"2019-05-21 11:43:15.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-05-21 12:22:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-21 14:27:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a65d5dbca7a14fd75938113b8df87fc37e26528","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 14:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4c6588fe725a1b1f796fc8b59ce9ca5cd0cea9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:06:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/41/660141/1/check/build-openstack-sphinx-docs/065f653/html/ : SUCCESS in 2m 55s\n- openstack-ansible-linters http://logs.openstack.org/41/660141/1/check/openstack-ansible-linters/db4265c/ : FAILURE in 4m 14s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/660141/1/check/openstack-ansible-functional-centos-7/276a978/ : FAILURE in 6m 08s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/41/660141/1/check/openstack-ansible-functional-centos-7-ssl/d9de6e2/ : FAILURE in 5m 42s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/41/660141/1/check/openstack-ansible-functional-opensuse-423/3ddd445/ : FAILURE in 4m 46s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/41/660141/1/check/openstack-ansible-functional-opensuse-423-ssl/7040362/ : FAILURE in 4m 31s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/41/660141/1/check/openstack-ansible-functional-ubuntu-xenial/0bcd75d/ : FAILURE in 4m 25s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/41/660141/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/bfd72f2/ : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e7a0e3b7db0b15e06aa0d6a0abbc5734eec3ef6c","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 16:32:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2ab457e7e398f638d12853576682913378d5972a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-20 16:34:08.000000000","message":"Patch Set 2:\n\nDidn\u0027t we vendor in the keys?","accounts_in_message":[],"_revision_number":2},{"id":"d87cfbdbebb7d32ddb981756a9e2d8afdf60de7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 16:52:11.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"9d9faa02337d2a378d48c84d2a7670cdbcf80b1b","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 16:52:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ae32a472245867f8e26d9318abe2881646487ae9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:18:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/41/660141/2/check/build-openstack-sphinx-docs/2ffd1f1/html/ : SUCCESS in 3m 09s\n- openstack-ansible-linters http://logs.openstack.org/41/660141/2/check/openstack-ansible-linters/d12df20/ : FAILURE in 5m 57s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/660141/2/check/openstack-ansible-functional-centos-7/5a41d44/ : FAILURE in 5m 07s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/41/660141/2/check/openstack-ansible-functional-centos-7-ssl/dcaffb9/ : FAILURE in 5m 38s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/41/660141/2/check/openstack-ansible-functional-opensuse-423/a2a2a98/ : FAILURE in 4m 08s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/41/660141/2/check/openstack-ansible-functional-opensuse-423-ssl/56c7ac8/ : FAILURE in 4m 54s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/41/660141/2/check/openstack-ansible-functional-ubuntu-xenial/4d9277f/ : FAILURE in 4m 07s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/41/660141/2/check/openstack-ansible-functional-ubuntu-xenial-ssl/f6db02d/ : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d2e0180077206318705b5c024abbb048655190f7","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 17:37:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"af74e00071c4cf9ce576e6b4d6f69dbdbc671079","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:50:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/41/660141/3/check/build-openstack-sphinx-docs/efac46d/html/ : SUCCESS in 3m 14s\n- openstack-ansible-linters http://logs.openstack.org/41/660141/3/check/openstack-ansible-linters/9c7333a/ : FAILURE in 4m 38s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/660141/3/check/openstack-ansible-functional-centos-7/ac496d0/ : FAILURE in 5m 02s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/41/660141/3/check/openstack-ansible-functional-centos-7-ssl/11e4f3d/ : FAILURE in 5m 20s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/41/660141/3/check/openstack-ansible-functional-opensuse-423/2c927c1/ : FAILURE in 3m 58s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/41/660141/3/check/openstack-ansible-functional-opensuse-423-ssl/56e7ee1/ : FAILURE in 3m 59s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/41/660141/3/check/openstack-ansible-functional-ubuntu-xenial/20518f1/ : FAILURE in 4m 26s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/41/660141/3/check/openstack-ansible-functional-ubuntu-xenial-ssl/a10a261/ : FAILURE in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"bfada2bc868d5b28b0a8d239e8893d0fcef94044","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 18:26:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7eb7f7cdf654d8c9dcb71029d41c6173b440139c","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 18:40:19.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"629e5d303974295eb992787def4c9f8f46c6bc98","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-20 19:02:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"769b29a1df54d7efdb8d0e98f2edd5e222ed70ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:39:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/41/660141/6/check/build-openstack-sphinx-docs/56a849d/html/ : SUCCESS in 3m 57s\n- openstack-ansible-linters http://logs.openstack.org/41/660141/6/check/openstack-ansible-linters/c07f0f8/ : SUCCESS in 6m 26s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/660141/6/check/openstack-ansible-functional-centos-7/d3ecfa2/ : SUCCESS in 29m 49s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/41/660141/6/check/openstack-ansible-functional-centos-7-ssl/4ff65e1/ : SUCCESS in 28m 36s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/41/660141/6/check/openstack-ansible-functional-ubuntu-xenial/a3680d7/ : SUCCESS in 16m 37s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/41/660141/6/check/openstack-ansible-functional-ubuntu-xenial-ssl/9e4f7b2/ : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":6},{"id":"9d509c4bc05b6c4729a71745743c03c2ac8e14c0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 21:27:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"6702b4e67ad45e2fa86874969105429902919434","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-20 21:50:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9036cc53a7a2d54e2d774f98d66c42a1b46a7009","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 11:43:15.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f6e6fcaad124086660588e698600e913c73cf704","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 12:22:59.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ddb0f29da48a5ea08689a3b6c99cf3fbe30c810f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 13:49:16.000000000","message":"Patch Set 6: Workflow+1\n\ngate please","accounts_in_message":[],"_revision_number":6},{"id":"a4f91fe5e2f07112d849be7bc4794bd2ad3bd025","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 13:49:51.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"f59d152f440d8647613116bc3be72974b5b915e1","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 13:50:09.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"40e602ee1c6bda6b4c7934758801f9d8859e1018","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 13:50:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b9886a38e25c257e23b01b6de1dbda0c93c8515d","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-21 13:51:06.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"53c026613feb84971353699ab208b86b5dde9c5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:27:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/41/660141/6/gate/build-openstack-sphinx-docs/3b29976/html/ : SUCCESS in 2m 58s\n- openstack-ansible-linters http://logs.openstack.org/41/660141/6/gate/openstack-ansible-linters/1e54bc6/ : SUCCESS in 6m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/41/660141/6/gate/openstack-ansible-functional-centos-7/6842774/ : SUCCESS in 32m 23s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/41/660141/6/gate/openstack-ansible-functional-centos-7-ssl/19bf323/ : SUCCESS in 30m 07s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/41/660141/6/gate/openstack-ansible-functional-ubuntu-xenial/e0515a4/ : SUCCESS in 16m 28s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/41/660141/6/gate/openstack-ansible-functional-ubuntu-xenial-ssl/eaa8944/ : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":6},{"id":"f13f872ef3fe113bf4468d003ec0f09fb23879c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:27:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5e3e8327d5d709a70b1e105154299be59d2575e0","revisions":{"bbc22ba7d6aea04d56d404351fe55463606831a5":{"kind":"REWORK","_number":1,"created":"2019-05-20 14:52:45.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/41/660141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/660141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/1"}}},"commit":{"parents":[{"commit":"80c8f8ba9d2dda1864982f64c5aef625fffca50d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80c8f8ba9d2dda1864982f64c5aef625fffca50d"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bbc22ba7d6aea04d56d404351fe55463606831a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/bbc22ba7d6aea04d56d404351fe55463606831a5"}]},"branch":"refs/heads/stable/queens"},"5e147661c67572095a3519d8e8b4349773526de5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-20 16:32:33.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/41/660141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/660141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/2"}}},"commit":{"parents":[{"commit":"80c8f8ba9d2dda1864982f64c5aef625fffca50d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80c8f8ba9d2dda1864982f64c5aef625fffca50d"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 16:32:31.000000000","tz":0},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nDepends-On: I67ebe9e813c95a3a8bcf204678ddccbcea7a45cd\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5e147661c67572095a3519d8e8b4349773526de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5e147661c67572095a3519d8e8b4349773526de5"}]},"branch":"refs/heads/stable/queens"},"d1a61aad08fded5e53d29ed78c3fcddabd859f8d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-20 17:37:30.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/41/660141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/660141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/3"}}},"commit":{"parents":[{"commit":"80c8f8ba9d2dda1864982f64c5aef625fffca50d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80c8f8ba9d2dda1864982f64c5aef625fffca50d"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 17:37:29.000000000","tz":0},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nDepends-On: I746b9dbb090dd484a36e826f06beda1eba3f55f4\nDepends-On: I67ebe9e813c95a3a8bcf204678ddccbcea7a45cd\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d1a61aad08fded5e53d29ed78c3fcddabd859f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/d1a61aad08fded5e53d29ed78c3fcddabd859f8d"}]},"branch":"refs/heads/stable/queens"},"6c58ce0a8341a02afd273df099abbed99465b661":{"kind":"REWORK","_number":4,"created":"2019-05-20 18:26:27.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/41/660141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/660141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/4"}}},"commit":{"parents":[{"commit":"80c8f8ba9d2dda1864982f64c5aef625fffca50d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80c8f8ba9d2dda1864982f64c5aef625fffca50d"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 18:26:09.000000000","tz":-300},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\nAlso pull in the updated tests-repo-clone script so the correct tests\nrepo path is used.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6c58ce0a8341a02afd273df099abbed99465b661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6c58ce0a8341a02afd273df099abbed99465b661"}]},"branch":"refs/heads/stable/queens"},"ed03aef47f097156f773162ca5341477a12e9ad6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-05-20 18:40:19.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/41/660141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/660141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/5"}}},"commit":{"parents":[{"commit":"80c8f8ba9d2dda1864982f64c5aef625fffca50d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80c8f8ba9d2dda1864982f64c5aef625fffca50d"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 18:40:10.000000000","tz":-300},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\nAlso pull in the updated tests-repo-clone script so the correct tests\nrepo path is used.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nDepends-On: I67ebe9e813c95a3a8bcf204678ddccbcea7a45cd\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ed03aef47f097156f773162ca5341477a12e9ad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ed03aef47f097156f773162ca5341477a12e9ad6"}]},"branch":"refs/heads/stable/queens"},"5e3e8327d5d709a70b1e105154299be59d2575e0":{"kind":"REWORK","_number":6,"created":"2019-05-20 19:02:55.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/41/660141/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/41/660141/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/41/660141/6"}}},"commit":{"parents":[{"commit":"80c8f8ba9d2dda1864982f64c5aef625fffca50d","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/80c8f8ba9d2dda1864982f64c5aef625fffca50d"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 19:02:27.000000000","tz":-300},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\nAlso pull in the updated tests-repo-clone script so the correct tests\nrepo path is used.\n\nopenSUSE 42.3 jobs are also removed since 42.3 is EOL soon and the jobs\nare broken since they leverage LXC containers.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nDepends-On: I67ebe9e813c95a3a8bcf204678ddccbcea7a45cd\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5e3e8327d5d709a70b1e105154299be59d2575e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5e3e8327d5d709a70b1e105154299be59d2575e0"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
