)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~662107","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Fpike~I9fb16b7f046530efc124e760bc7870c13b4e8d32","project":"openstack/openstack-ansible-galera_server","branch":"stable/pike","topic":"percona-release-pike","hashtags":[],"change_id":"I9fb16b7f046530efc124e760bc7870c13b4e8d32","subject":"Update percona-release package","status":"MERGED","created":"2019-05-29 22:34:19.000000000","updated":"2019-06-05 17:39:02.000000000","submitted":"2019-06-05 17:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662107-1559756342078-3b7fc237","meta_rev_id":"56e0c68866faf566398fd71dc3bdac4266b73147","_number":662107,"virtual_id_number":662107,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-30 20:16:22.000000000","_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-05 17:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2019-05-30 22:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-05 16:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"date":"2019-06-05 15:03:18.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-05 16:54:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-30 22:33:05.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-05 15:03:18.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-05 16:54:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-05 17:39:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a3513ade99d80f5834c00a9b0e21fa9afd80450","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-29 22:34:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ebdd90307044e6b12858d51308098dfbbfd436e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 23:36:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/07/662107/1/check/build-openstack-sphinx-docs/f7fee47/html/ : SUCCESS in 3m 37s\n- openstack-ansible-linters http://logs.openstack.org/07/662107/1/check/openstack-ansible-linters/36c76a4/ : FAILURE in 5m 43s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/07/662107/1/check/openstack-ansible-functional-centos-7/2ae8b30/ : FAILURE in 6m 13s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/07/662107/1/check/openstack-ansible-functional-ubuntu-xenial/dba3a83/ : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e29ddb85b71f3cd3508ccd164ae992278d92bf62","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-05-30 20:16:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4106430aea62cda83ae67b3add2fa301daee07a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 21:07:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/07/662107/1/check/build-openstack-sphinx-docs/d7e8352/html/ : SUCCESS in 3m 36s\n- openstack-ansible-linters http://logs.openstack.org/07/662107/1/check/openstack-ansible-linters/817d518/ : SUCCESS in 4m 27s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/07/662107/1/check/openstack-ansible-functional-centos-7/3080fdb/ : SUCCESS in 30m 48s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/07/662107/1/check/openstack-ansible-functional-ubuntu-xenial/4f62b2c/ : SUCCESS in 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"50adf0fbf848b7a9421f8b019c65f0c722b74503","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-30 22:33:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"66d9a8fce309039bfa4d998047d12c93e05d148d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-05 16:54:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7b49fce474820ed36de18fc956af1da79ed325a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 16:54:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65c32813d9073aa04e584323ab30cd9c0f229e52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 17:39:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/07/662107/1/gate/build-openstack-sphinx-docs/e3efcb0/html/ : SUCCESS in 3m 38s\n- openstack-ansible-linters http://logs.openstack.org/07/662107/1/gate/openstack-ansible-linters/8feae06/ : SUCCESS in 6m 06s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/07/662107/1/gate/openstack-ansible-functional-centos-7/c2c6549/ : SUCCESS in 33m 51s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/07/662107/1/gate/openstack-ansible-functional-ubuntu-xenial/6020fd2/ : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1ae055ad7277e7d805a1cdbe4d1a78c7e04171a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 17:39:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f8992758b54eaa8b9967e94332584660c7dca92","revisions":{"1f8992758b54eaa8b9967e94332584660c7dca92":{"kind":"REWORK","_number":1,"created":"2019-05-29 22:34:19.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/07/662107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/07/662107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/662107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/662107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/662107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/07/662107/1"}}},"commit":{"parents":[{"commit":"47916da9f6359810350d2d37c35644852b819f43","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/47916da9f6359810350d2d37c35644852b819f43"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-20 14:51:58.000000000","tz":-300},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2019-05-29 22:30:13.000000000","tz":-300},"subject":"Update percona-release package","message":"Update percona-release package\n\nIn February 2019, Percona updated their RPM signing keys[1], requiring\nthe use of an updated percona-release package to install the new\nkeys.\n\nAlso pull in the updated tests-repo-clone script so the correct tests\nrepo path is used.\n\nopenSUSE 42.3 jobs are also removed since 42.3 is EOL soon and the jobs\nare broken since they leverage LXC containers.\n\n[1] https://www.percona.com/blog/2019/02/05/new-percona-package-signing-key-requires-update-on-rhel-and-centos/\n\nDepends-On: I67ebe9e813c95a3a8bcf204678ddccbcea7a45cd\nChange-Id: I9fb16b7f046530efc124e760bc7870c13b4e8d32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1f8992758b54eaa8b9967e94332584660c7dca92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1f8992758b54eaa8b9967e94332584660c7dca92"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
