)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~662338","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I0e530e67238a42e63f25ce0d59eefca2b7a2099a","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I0e530e67238a42e63f25ce0d59eefca2b7a2099a","subject":"Use a private namespaced var for cluster name check","status":"MERGED","created":"2019-05-30 21:27:29.000000000","updated":"2019-07-30 20:11:20.000000000","submitted":"2019-07-30 20:11:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662338-1564517480778-c60208eb","meta_rev_id":"ecd5d7408e24fd167f38007500be097253d5b5af","_number":662338,"virtual_id_number":662338,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-04 16:29:57.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 20:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-07-30 17:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-30 18:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-30 18:41:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 17:16:27.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-30 18:41:15.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-30 20:11:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25d78740db3a3fd5c9b0d026d98d7fc30f0c8547","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-30 21:27:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8494d1aeb85247c3453692ae33800236c5e3482a","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-30 21:28:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4674dc5d77f7ed7a1f6d5af85890353766291442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 22:55:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/38/662338/2/check/openstack-ansible-linters/c35092b/ : SUCCESS in 5m 28s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-debian-stable/f8a3585/ : SUCCESS in 16m 12s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-centos-7/a430239/ : SUCCESS in 19m 39s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-bionic/b9472d7/ : SUCCESS in 18m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-debian-stable/f5aa22c/ : SUCCESS in 1h 11m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-centos-7/3a85b6a/ : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/e752862/ : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bcb9cf2/ : SUCCESS in 58m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e184716/ : SUCCESS in 1h 03m 28s\n- openstack-tox-docs http://logs.openstack.org/38/662338/2/check/openstack-tox-docs/efbd7e7/html/ : SUCCESS in 2m 37s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-centos-7-ssl/7a65b70/ : SUCCESS in 21m 56s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-xenial-ssl/fe966ea/ : FAILURE in 9m 39s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/1de3974/ : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c766c5c952e4874383f015f26c3e2fb9ef4b9c13","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-04 16:29:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"671574e6887dc7efd66484f36ee9e912dd3d52da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 17:59:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/38/662338/2/check/openstack-ansible-linters/bbb7548/ : FAILURE in 5m 48s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-debian-stable/63fb5c0/ : FAILURE in 5m 44s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-centos-7/65e469e/ : FAILURE in 6m 46s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-bionic/c6e5cfd/ : SUCCESS in 15m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-debian-stable/01f668a/ : SUCCESS in 1h 03m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-centos-7/6f65995/ : SUCCESS in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/b99d61e/ : SUCCESS in 1h 06m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2e258b2/ : SUCCESS in 57m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/bad870a/ : SUCCESS in 58m 06s\n- openstack-tox-docs http://logs.openstack.org/38/662338/2/check/openstack-tox-docs/f339782/html/ : SUCCESS in 2m 49s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-centos-7-ssl/37801b4/ : FAILURE in 6m 42s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-xenial-ssl/c3f8f64/ : FAILURE in 9m 55s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/44d531d/ : SUCCESS in 19m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f85e8347ca4af4878368b91e09b59585d94decca","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-27 21:38:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"df7c8f11ce5e5ec45ea9680553af7303a3ac564c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 23:13:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/38/662338/2/check/openstack-ansible-linters/a2ae96c/ : SUCCESS in 4m 49s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-debian-stable/faa6908/ : SUCCESS in 16m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-centos-7/94c11bd/ : SUCCESS in 20m 00s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-bionic/f087cd1/ : SUCCESS in 15m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-debian-stable/891b295/ : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-centos-7/ec26c9f/ : SUCCESS in 1h 29m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/88cc245/ : SUCCESS in 1h 02m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/33020ee/ : SUCCESS in 53m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/2a5a99e/ : SUCCESS in 58m 26s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/816301a/ : FAILURE in 40m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/38/662338/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/e9f2db0/ : SUCCESS in 43m 54s\n- openstack-tox-docs http://logs.openstack.org/38/662338/2/check/openstack-tox-docs/952a1da/html/ : SUCCESS in 2m 58s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-centos-7-ssl/2ed4453/ : SUCCESS in 17m 04s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-opensuse-15-ssl/f4d9ba4/ : FAILURE in 29m 41s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/38/662338/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/cfcbe43/ : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e0a0b18a1780b8d87c671e8a2adef429537c1a62","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 17:16:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"95fc657c54feec2d41da38bb302061a5955a2d8b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-30 18:41:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fd797ca09b85be8e2dcde7bb7ba5625ba074fd68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 18:41:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1e4f884c34db7b5e343c450c327e2b023388c86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:11:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ecd5d7408e24fd167f38007500be097253d5b5af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:11:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/38/662338/2/gate/openstack-ansible-linters/3af7718/ : SUCCESS in 5m 27s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/38/662338/2/gate/openstack-ansible-functional-debian-stable/263824e/ : SUCCESS in 16m 37s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/38/662338/2/gate/openstack-ansible-functional-centos-7/5866659/ : SUCCESS in 20m 16s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/38/662338/2/gate/openstack-ansible-functional-ubuntu-bionic/5b204f4/ : SUCCESS in 15m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/38/662338/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/a20500b/ : SUCCESS in 59m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/38/662338/2/gate/openstack-ansible-deploy-aio_metal-centos-7/9a00088/ : SUCCESS in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/38/662338/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/22d7f44/ : SUCCESS in 1h 14m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/38/662338/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/aba426e/ : SUCCESS in 1h 07m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/38/662338/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/97b8d35/ : SUCCESS in 58m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/38/662338/2/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/607d32d/ : SUCCESS in 42m 30s\n- openstack-tox-docs http://logs.openstack.org/38/662338/2/gate/openstack-tox-docs/24b7e98/html/ : SUCCESS in 2m 34s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/38/662338/2/gate/openstack-ansible-functional-centos-7-ssl/4a5ec9d/ : SUCCESS in 18m 38s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/38/662338/2/gate/openstack-ansible-functional-ubuntu-bionic-ssl/6185c75/ : SUCCESS in 18m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13598110846ce8e34c7e59c90ef8aff0539f99af","revisions":{"c8b9225c8bd7c641e002003b0305206ea4090d7c":{"kind":"REWORK","_number":1,"created":"2019-05-30 21:27:29.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/38/662338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/38/662338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/1"}}},"commit":{"parents":[{"commit":"24601204275798238779fd3b2605153a07123bd8","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/24601204275798238779fd3b2605153a07123bd8"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-30 21:26:00.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-30 21:26:00.000000000","tz":-300},"subject":"Use a private namespaced var for cluster name check","message":"Use a private namespaced var for cluster name check\n\nThe galera cluster check uses \"cluster_name\" as a task return variable\nwhen pulling the cluster name from an existing cluster. As one\nmight imagine, the cluster_name variable is very ambiguous and could\nexist in a user\u0027s user_variables or inventory somewhere, which breaks\nthis check and fails the role execution. Instead of using \u0027cluster_name\u0027\nwe can just return the task result to a private namespaced var\n\u0027_cluster_name\u0027 and use that for the assertion instead.\n\nChange-Id: I0e530e67238a42e63f25ce0d59eefca2b7a2099a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c8b9225c8bd7c641e002003b0305206ea4090d7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/c8b9225c8bd7c641e002003b0305206ea4090d7c"}]},"branch":"refs/heads/master"},"13598110846ce8e34c7e59c90ef8aff0539f99af":{"kind":"REWORK","_number":2,"created":"2019-05-30 21:28:32.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/38/662338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/38/662338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/38/662338/2"}}},"commit":{"parents":[{"commit":"24601204275798238779fd3b2605153a07123bd8","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/24601204275798238779fd3b2605153a07123bd8"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-30 21:26:00.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-05-30 21:28:31.000000000","tz":0},"subject":"Use a private namespaced var for cluster name check","message":"Use a private namespaced var for cluster name check\n\nThe galera cluster check uses \"cluster_name\" as a task return variable\nwhen pulling the cluster name from an existing cluster. As one\nmight imagine, the cluster_name variable is very ambiguous and could\nexist in a user\u0027s user_variables or inventory somewhere, which breaks\nthis check and fails the role execution. Instead of using \u0027cluster_name\u0027\nwe can just return the task result to a private namespaced var\n\u0027_cluster_name\u0027 and use that for the assertion instead.\n\n\u0027node_status\u0027 is a similar check, also changed to \u0027_node_status\u0027\n\nChange-Id: I0e530e67238a42e63f25ce0d59eefca2b7a2099a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/13598110846ce8e34c7e59c90ef8aff0539f99af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/13598110846ce8e34c7e59c90ef8aff0539f99af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
