)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~664287","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Frocky~Ib4a03cff210f50f1545478e858b80d155ddce18e","project":"openstack/openstack-ansible-galera_server","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ib4a03cff210f50f1545478e858b80d155ddce18e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:43:17.000000000","updated":"2019-06-10 18:27:35.000000000","submitted":"2019-06-10 18:27:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664287-1560191255836-b5d80b2d","meta_rev_id":"d4d746df0f0008f0f3f69cb21af078b11536ef3f","_number":664287,"virtual_id_number":664287,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 18:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:31:14.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 18:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bdc5b27200ce7a60ce198f51d04e593db77fa38","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:43:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c14db0c5cf2800390a6204e21d5e3ff1a718d77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:34:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/664287/1/check/openstack-ansible-linters/de5e1cc/ : SUCCESS in 5m 54s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-centos-7/f4ae384/ : SUCCESS in 22m 32s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-opensuse-423/cfe6fb8/ : FAILURE in 6m 28s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-opensuse-150/14ed822/ : FAILURE in 8m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-ubuntu-bionic/2288e02/ : SUCCESS in 17m 19s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-ubuntu-xenial/f1e9b31/ : SUCCESS in 15m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/664287/1/check/build-openstack-sphinx-docs/6c78fcb/html/ : SUCCESS in 3m 48s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-centos-7-ssl/cfe182b/ : SUCCESS in 19m 51s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-opensuse-423-ssl/f70df09/ : FAILURE in 7m 51s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/76d225d/ : SUCCESS in 18m 21s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/87/664287/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/1af1565/ : SUCCESS in 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bffe15f5222e70525313e99b1074a5beb9c93f2e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:31:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31e28ed2df65438d9a3c5e3d4545ca8ac65adbac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:31:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e239e8588fb1fa776597f0272f2de79e80e9d106","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:27:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e2234ad7721b1306931e476374b405178e880b7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:27:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/664287/1/gate/openstack-ansible-linters/79f8829/ : SUCCESS in 4m 35s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-centos-7/a4173e0/ : SUCCESS in 26m 01s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-opensuse-423/4d1a87b/ : FAILURE in 8m 27s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-ubuntu-bionic/1809a10/ : SUCCESS in 18m 55s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-ubuntu-xenial/1d96abd/ : SUCCESS in 17m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/664287/1/gate/build-openstack-sphinx-docs/bd3afab/html/ : SUCCESS in 3m 17s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-centos-7-ssl/a915c57/ : SUCCESS in 20m 46s\n- openstack-ansible-functional-opensuse-423-ssl http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-opensuse-423-ssl/5161997/ : FAILURE in 7m 14s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-ubuntu-xenial-ssl/ac7e220/ : SUCCESS in 17m 35s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/87/664287/1/gate/openstack-ansible-functional-ubuntu-bionic-ssl/7fc8d62/ : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f551e6a51d6b4e80bca95e2dd030103883ce438a","revisions":{"f551e6a51d6b4e80bca95e2dd030103883ce438a":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:43:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/664287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/87/664287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/664287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/664287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/664287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/87/664287/1"}}},"commit":{"parents":[{"commit":"6329358505e21f33e8f348cdb185b1ddbb8c3e31","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/6329358505e21f33e8f348cdb185b1ddbb8c3e31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:43:16.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ib4a03cff210f50f1545478e858b80d155ddce18e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f551e6a51d6b4e80bca95e2dd030103883ce438a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f551e6a51d6b4e80bca95e2dd030103883ce438a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
