)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~666272","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ia0985c03d51c3bfe06b98ae73a3533c3402453e3","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Ia0985c03d51c3bfe06b98ae73a3533c3402453e3","subject":"Fix galera_server installation for openSUSE Leap 15.1","status":"MERGED","created":"2019-06-19 12:12:32.000000000","updated":"2019-06-26 23:42:43.000000000","submitted":"2019-06-26 23:42:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666272-1561592563681-16d569af","meta_rev_id":"80eeb7cb67e4257f412b795c9c386b937769ac27","_number":666272,"virtual_id_number":666272,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-26 23:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-26 17:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-26 22:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-26 22:11:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 09:25:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-26 22:11:59.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-26 23:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21a9146e8b9a5179954b0d7ed5c6ac023ece7a4a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-19 12:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b832015385e35a0bc9f0d5b1edfdefbf121428aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:40:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/72/666272/1/check/openstack-ansible-linters/823294e/ : SUCCESS in 5m 17s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/666272/1/check/openstack-ansible-functional-debian-stable/264d91e/ : SUCCESS in 15m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/666272/1/check/openstack-ansible-functional-centos-7/aff34c6/ : SUCCESS in 20m 59s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/666272/1/check/openstack-ansible-functional-ubuntu-bionic/08f521c/ : SUCCESS in 21m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/72/666272/1/check/openstack-ansible-deploy-aio_metal-debian-stable/e3e078c/ : SUCCESS in 1h 01m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/72/666272/1/check/openstack-ansible-deploy-aio_metal-centos-7/4f7df0c/ : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/72/666272/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/db1d38d/ : FAILURE in 17m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/72/666272/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dd18716/ : SUCCESS in 58m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/72/666272/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c9e2643/ : SUCCESS in 1h 06m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/72/666272/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/1e11cce/ : FAILURE in 13m 27s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/72/666272/1/check/openstack-tox-docs/9cdbe75/html/ : SUCCESS in 4m 09s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/72/666272/1/check/openstack-ansible-functional-centos-7-ssl/dcda4ca/ : SUCCESS in 19m 22s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/72/666272/1/check/openstack-ansible-functional-ubuntu-xenial-ssl/0a8eaba/ : FAILURE in 10m 24s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/72/666272/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/7078e75/ : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2bd3a1d4519cd9244f40ffbbf74ca43ccfe5568f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-25 09:21:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"604d6417e5049f4bbeecad197e5961548fb4a617","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-25 09:25:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8551e2fe1b3da3cf45bb017b1c44e73fbd219472","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 12:14:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/72/666272/2/check/openstack-ansible-linters/e2a77d9/ : SUCCESS in 5m 45s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/666272/2/check/openstack-ansible-functional-debian-stable/8bb355e/ : SUCCESS in 16m 05s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/666272/2/check/openstack-ansible-functional-centos-7/a2dd4e2/ : SUCCESS in 18m 32s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/666272/2/check/openstack-ansible-functional-ubuntu-bionic/86b8a93/ : SUCCESS in 15m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable finger://ze07.openstack.org/c39c18a4aeff4b07b4e9539492d7bb32 : RETRY_LIMIT in 49m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 finger://ze10.openstack.org/942142cf01e4465a8e84f5b1418ad85e : RETRY_LIMIT in 55m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/72/666272/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/4261b7c/ : POST_FAILURE in 33m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic finger://ze12.openstack.org/8ad45d651b8d4e2fbc95b48cbe6d5766 : RETRY_LIMIT in 59m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 finger://ze10.openstack.org/db23e2f5bc7e4c4581e7b29f3cd05393 : RETRY_LIMIT in 55m 28s\n- openstack-tox-docs http://logs.openstack.org/72/666272/2/check/openstack-tox-docs/a18999f/html/ : SUCCESS in 3m 49s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/72/666272/2/check/openstack-ansible-functional-centos-7-ssl/52248e3/ : SUCCESS in 27m 24s\n- openstack-ansible-functional-ubuntu-xenial-ssl http://logs.openstack.org/72/666272/2/check/openstack-ansible-functional-ubuntu-xenial-ssl/b6e362d/ : FAILURE in 21m 39s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/72/666272/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/b639b8a/ : SUCCESS in 22m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0175de236a8530cd6681983028f918fdf34106ea","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-26 16:53:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46f8c1095ec59315fd1aa17074634a917868e05b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-26 17:00:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6731118f2848d2d7f4f2bb20ef4f916ab0af6a4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:48:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/72/666272/3/check/openstack-ansible-linters/c3d3814/ : SUCCESS in 5m 05s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/666272/3/check/openstack-ansible-functional-debian-stable/9ae0cd5/ : SUCCESS in 15m 28s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/666272/3/check/openstack-ansible-functional-centos-7/84f2293/ : SUCCESS in 22m 54s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/666272/3/check/openstack-ansible-functional-ubuntu-bionic/01ddbb4/ : SUCCESS in 23m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/72/666272/3/check/openstack-ansible-deploy-aio_metal-debian-stable/e2fdba5/ : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/72/666272/3/check/openstack-ansible-deploy-aio_metal-centos-7/620d68f/ : SUCCESS in 1h 13m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/72/666272/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/3f60dd9/ : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/72/666272/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2301036/ : SUCCESS in 56m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/72/666272/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/93e5331/ : SUCCESS in 1h 10m 48s\n- openstack-tox-docs http://logs.openstack.org/72/666272/3/check/openstack-tox-docs/4ca5271/html/ : SUCCESS in 2m 42s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/72/666272/3/check/openstack-ansible-functional-centos-7-ssl/1b8ee16/ : SUCCESS in 23m 29s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/72/666272/3/check/openstack-ansible-functional-ubuntu-bionic-ssl/3a80677/ : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":3},{"id":"382a3859bb0cbed0ba4f9376310e5ca55b26bff9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-26 22:11:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e09190018d8bef1d2a807db732e37008b8914f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:12:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bcc7abc0649961052b518a5e80b361af5a512c8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 23:42:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/72/666272/3/gate/openstack-ansible-linters/959f2ed/ : SUCCESS in 4m 59s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/72/666272/3/gate/openstack-ansible-functional-debian-stable/08b8e60/ : SUCCESS in 16m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/72/666272/3/gate/openstack-ansible-functional-centos-7/bb86975/ : SUCCESS in 17m 57s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/72/666272/3/gate/openstack-ansible-functional-ubuntu-bionic/78691a1/ : SUCCESS in 16m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/72/666272/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/310e167/ : SUCCESS in 1h 03m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/72/666272/3/gate/openstack-ansible-deploy-aio_metal-centos-7/d771466/ : SUCCESS in 1h 22m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/72/666272/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/4c281a2/ : SUCCESS in 1h 11m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/72/666272/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a66e6ab/ : SUCCESS in 58m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/72/666272/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/fb7f9bb/ : SUCCESS in 1h 07m 05s\n- openstack-tox-docs http://logs.openstack.org/72/666272/3/gate/openstack-tox-docs/325d7f7/html/ : SUCCESS in 2m 55s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/72/666272/3/gate/openstack-ansible-functional-centos-7-ssl/577b7e6/ : SUCCESS in 21m 19s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/72/666272/3/gate/openstack-ansible-functional-ubuntu-bionic-ssl/d89a49f/ : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":3},{"id":"80eeb7cb67e4257f412b795c9c386b937769ac27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 23:42:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5965bf6e5021e71442e2abb8c7ff52a868259c52","revisions":{"996feeed372ce6c32e8c74a3df509199397268bc":{"kind":"REWORK","_number":1,"created":"2019-06-19 12:12:32.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/72/666272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/72/666272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/1"}}},"commit":{"parents":[{"commit":"24601204275798238779fd3b2605153a07123bd8","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/24601204275798238779fd3b2605153a07123bd8"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 12:11:45.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 12:12:25.000000000","tz":120},"subject":"Fix galera_server installation for openSUSE Leap 15.1","message":"Fix galera_server installation for openSUSE Leap 15.1\n\nThe MariaDB repositories are offering a download repo under\nthe leap major version, e.g. \"opensuse15\" or \"opensuse42\" for all leap\n42.x and leap 15.x versions, respectively. There isn\u0027t a specific\nbuild for openSUSE Leap 15.1, so the previous logic selected\na repo url of opensuse-151 which does not exist.\n\nDepends-On: https://review.opendev.org/666058\nChange-Id: Ia0985c03d51c3bfe06b98ae73a3533c3402453e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/996feeed372ce6c32e8c74a3df509199397268bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/996feeed372ce6c32e8c74a3df509199397268bc"}]},"branch":"refs/heads/master"},"7aa3e59c2ec98332eb6953e06e6858a290100a38":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-25 09:21:32.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/72/666272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/72/666272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/2"}}},"commit":{"parents":[{"commit":"24601204275798238779fd3b2605153a07123bd8","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/24601204275798238779fd3b2605153a07123bd8"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 12:11:45.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-25 09:21:20.000000000","tz":120},"subject":"Fix galera_server installation for openSUSE Leap 15.1","message":"Fix galera_server installation for openSUSE Leap 15.1\n\nThe MariaDB repositories are offering a download repo under\nthe leap major version, e.g. \"opensuse15\" or \"opensuse42\" for all leap\n42.x and leap 15.x versions, respectively. There isn\u0027t a specific\nbuild for openSUSE Leap 15.1, so the previous logic selected\na repo url of opensuse-151 which does not exist.\n\nChange-Id: Ia0985c03d51c3bfe06b98ae73a3533c3402453e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7aa3e59c2ec98332eb6953e06e6858a290100a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7aa3e59c2ec98332eb6953e06e6858a290100a38"}]},"branch":"refs/heads/master"},"5965bf6e5021e71442e2abb8c7ff52a868259c52":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-26 17:00:28.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/72/666272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/72/666272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/72/666272/3"}}},"commit":{"parents":[{"commit":"00eef359e290258f54a4b32fa0c73e7d42425e7a","subject":"Remove xenial-ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/00eef359e290258f54a4b32fa0c73e7d42425e7a"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-19 12:11:45.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-26 16:56:52.000000000","tz":120},"subject":"Fix galera_server installation for openSUSE Leap 15.1","message":"Fix galera_server installation for openSUSE Leap 15.1\n\nThe MariaDB repositories are offering a download repo under\nthe leap major version, e.g. \"opensuse15\" or \"opensuse42\" for all leap\n42.x and leap 15.x versions, respectively. There isn\u0027t a specific\nbuild for openSUSE Leap 15.1, so the previous logic selected\na repo url of opensuse-151 which does not exist.\n\nChange-Id: Ia0985c03d51c3bfe06b98ae73a3533c3402453e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5965bf6e5021e71442e2abb8c7ff52a868259c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5965bf6e5021e71442e2abb8c7ff52a868259c52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
