)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~667683","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I4cf89ee23e28ce20b4076d4eb81daa04a8055499","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I4cf89ee23e28ce20b4076d4eb81daa04a8055499","subject":"Replace opensuse 42.3 functional testing with opensuse-15","status":"MERGED","created":"2019-06-26 17:00:28.000000000","updated":"2019-06-27 16:16:49.000000000","submitted":"2019-06-27 16:16:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667683-1561652209223-a45a0009","meta_rev_id":"53b0eb7a14efeb9b9c7bb94f39dfc1cbc574949e","_number":667683,"virtual_id_number":667683,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-27 16:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 07:03:51.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-27 08:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-06-27 08:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 07:03:51.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-27 08:36:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-27 16:16:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72bd7d6c25a2334ce5b72a5df4633d6ffa568eba","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-26 17:00:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70b84d55038a53238e7e60ef656307ca9a9756bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:43:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/667683/1/check/openstack-ansible-linters/d484fe6/ : SUCCESS in 5m 24s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/667683/1/check/openstack-ansible-functional-debian-stable/b6eaa32/ : SUCCESS in 15m 23s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/667683/1/check/openstack-ansible-functional-centos-7/5c27b80/ : SUCCESS in 19m 18s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/667683/1/check/openstack-ansible-functional-ubuntu-bionic/d55169e/ : SUCCESS in 15m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/83/667683/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7c040f6/ : SUCCESS in 1h 03m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/83/667683/1/check/openstack-ansible-deploy-aio_metal-centos-7/108cbad/ : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/83/667683/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/98fa965/ : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/83/667683/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8ab6fbd/ : SUCCESS in 57m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/83/667683/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/00d9a0a/ : SUCCESS in 1h 03m 25s\n- openstack-tox-docs http://logs.openstack.org/83/667683/1/check/openstack-tox-docs/495721b/html/ : SUCCESS in 2m 44s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/83/667683/1/check/openstack-ansible-functional-centos-7-ssl/13427bb/ : SUCCESS in 26m 26s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/83/667683/1/check/openstack-ansible-functional-opensuse-15-ssl/3bea7c1/ : FAILURE in 35m 01s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/83/667683/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/29016ee/ : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fecce88a8a4e68578d9d7ac98eec81cf9e5b305b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-26 21:14:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2020a3f995a0b9a1da35da4774c4617d06265195","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 07:01:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"968e058f0c432efa9fb1017136954720923e8c72","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 07:01:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6b5fe7a94fa68a5bf5eb457eb884e47523eb1c8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 07:03:51.000000000","message":"Patch Set 2: Code-Review+2\n\nGood catch on this particular scenario.","accounts_in_message":[],"_revision_number":2},{"id":"38f2d3f4f86859cdcc0d8846d7f3ac1d577dde90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:28:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/667683/2/check/openstack-ansible-linters/df90442/ : SUCCESS in 5m 13s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-debian-stable/441f419/ : SUCCESS in 15m 18s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-centos-7/122c81b/ : SUCCESS in 19m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-ubuntu-bionic/7f48c76/ : SUCCESS in 16m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-debian-stable/0876953/ : SUCCESS in 1h 03m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-centos-7/2259713/ : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/1d6fb46/ : SUCCESS in 1h 20m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e5605f3/ : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/6b214b8/ : SUCCESS in 57m 53s\n- openstack-tox-docs http://logs.openstack.org/83/667683/2/check/openstack-tox-docs/1eb30d2/html/ : SUCCESS in 2m 48s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-centos-7-ssl/154f7b9/ : SUCCESS in 21m 22s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-opensuse-15-ssl/019cfcf/ : FAILURE in 26m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/8d1431a/ : SUCCESS in 16m 18s","accounts_in_message":[],"_revision_number":2},{"id":"dc63d3f7f9add7815cc571933516432fbd7e74c0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-27 08:36:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"61648b35e7c56924593cfcb107ca488c640d0648","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 08:36:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d9581fd9b0abbf41972e6c3571157a085b5e606","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 10:34:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/83/667683/2/gate/openstack-ansible-linters/8b4069a/ : SUCCESS in 10m 02s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-debian-stable/234fb34/ : FAILURE in 20m 08s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-centos-7/f486799/ : SUCCESS in 27m 49s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-ubuntu-bionic/a100f04/ : FAILURE in 16m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/2e1239f/ : SUCCESS in 1h 00m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-centos-7/4b146e7/ : RETRY_LIMIT in 11m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/c2013c0/ : SUCCESS in 1h 09m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ddb7aaa/ : SUCCESS in 59m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/4666290/ : SUCCESS in 1h 07m 51s\n- openstack-tox-docs http://logs.openstack.org/83/667683/2/gate/openstack-tox-docs/1ca1aa7/ : RETRY_LIMIT in 2m 28s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-centos-7-ssl/7e480f5/ : SUCCESS in 18m 15s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-ubuntu-bionic-ssl/2e21cd7/ : FAILURE in 16m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b1f1fe031f1e8193261db635aba5eac5f73150eb","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 12:57:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c24901557a65280d1b86e349effc7efd6bd960e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:29:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/667683/2/check/openstack-ansible-linters/c374417/ : SUCCESS in 6m 25s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-debian-stable/80f9028/ : SUCCESS in 17m 12s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-centos-7/f715d5e/ : SUCCESS in 20m 29s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-ubuntu-bionic/dfe061c/ : SUCCESS in 16m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-debian-stable/0dd9441/ : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-centos-7/c43085e/ : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/70d5d01/ : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/68f7cf7/ : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/83/667683/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9f344de/ : SUCCESS in 1h 02m 54s\n- openstack-tox-docs http://logs.openstack.org/83/667683/2/check/openstack-tox-docs/ef59fc1/html/ : SUCCESS in 2m 30s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-centos-7-ssl/8a1a873/ : SUCCESS in 18m 18s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-opensuse-15-ssl/a78bdbe/ : FAILURE in 31m 28s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/83/667683/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/4e18ac1/ : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6e9d4809856ec58e58811e81c3fbb9c994bd64ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:29:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b97bb5e7a7da6bc5e4ee76e791c9aa1644e3a0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:16:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/83/667683/2/gate/openstack-ansible-linters/c8a6165/ : SUCCESS in 5m 03s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-debian-stable/dc37e5c/ : SUCCESS in 18m 04s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-centos-7/4652f37/ : SUCCESS in 23m 08s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-ubuntu-bionic/5476405/ : SUCCESS in 15m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/176d538/ : SUCCESS in 1h 11m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-centos-7/c224d06/ : SUCCESS in 1h 16m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/b9ba6bc/ : SUCCESS in 1h 04m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2999635/ : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/83/667683/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/6b0d6eb/ : SUCCESS in 58m 08s\n- openstack-tox-docs http://logs.openstack.org/83/667683/2/gate/openstack-tox-docs/b120c62/html/ : SUCCESS in 2m 40s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-centos-7-ssl/68f4e87/ : SUCCESS in 18m 52s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/83/667683/2/gate/openstack-ansible-functional-ubuntu-bionic-ssl/d6a38d1/ : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":2},{"id":"53b0eb7a14efeb9b9c7bb94f39dfc1cbc574949e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:16:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"473d2050d43b539fc9dcccf5b5ca0e4e5f22d0a2","revisions":{"96b420334fa234c1f7f6b053fac9468c79b91233":{"kind":"REWORK","_number":1,"created":"2019-06-26 17:00:28.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/83/667683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/667683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/1"}}},"commit":{"parents":[{"commit":"5965bf6e5021e71442e2abb8c7ff52a868259c52","subject":"Fix galera_server installation for openSUSE Leap 15.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5965bf6e5021e71442e2abb8c7ff52a868259c52"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-26 16:59:21.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-26 16:59:21.000000000","tz":120},"subject":"Replace opensuse 42.3 functional testing with opensuse-15","message":"Replace opensuse 42.3 functional testing with opensuse-15\n\nopenSUSE Leap 42.3 is out of maintenance and end of life.\nIt has been superceded by Leap 15.0 and the current version\nas of today is 15.1. by using opensuse-15 we point to the\nlatest 15.x nodeset.\n\nChange-Id: I4cf89ee23e28ce20b4076d4eb81daa04a8055499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/96b420334fa234c1f7f6b053fac9468c79b91233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/96b420334fa234c1f7f6b053fac9468c79b91233"}]},"branch":"refs/heads/master"},"473d2050d43b539fc9dcccf5b5ca0e4e5f22d0a2":{"kind":"REWORK","_number":2,"created":"2019-06-27 07:01:50.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/83/667683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/83/667683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/83/667683/2"}}},"commit":{"parents":[{"commit":"5965bf6e5021e71442e2abb8c7ff52a868259c52","subject":"Fix galera_server installation for openSUSE Leap 15.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5965bf6e5021e71442e2abb8c7ff52a868259c52"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-26 16:59:21.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 07:01:08.000000000","tz":120},"subject":"Replace opensuse 42.3 functional testing with opensuse-15","message":"Replace opensuse 42.3 functional testing with opensuse-15\n\nopenSUSE Leap 42.3 is out of maintenance and end of life.\nIt has been superceded by Leap 15.0 and the current version\nas of today is 15.1. by using opensuse-15 we point to the\nlatest 15.x nodeset.\n\nChange-Id: I4cf89ee23e28ce20b4076d4eb81daa04a8055499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/473d2050d43b539fc9dcccf5b5ca0e4e5f22d0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/473d2050d43b539fc9dcccf5b5ca0e4e5f22d0a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
