)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~670403","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I4bb36e779002a6ce7a22fe308195de3b10aa1aba","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"I4bb36e779002a6ce7a22fe308195de3b10aa1aba","subject":"Add defaults for galera server and wrep cluster bind addresses","status":"MERGED","created":"2019-07-11 22:15:14.000000000","updated":"2020-06-24 13:32:31.000000000","submitted":"2020-06-16 20:16:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670403-1592338591610-ff50f346","meta_rev_id":"d22ca7ee0ebb893f711d300a487f3c01b03d48d3","_number":670403,"virtual_id_number":670403,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 20:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-06-16 14:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-06-16 07:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2020-06-16 14:25:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 15:13:58.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-06-15 19:36:02.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-16 07:55:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-16 14:25:40.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-16 20:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"942e700f672523c03ae8b0da7057a0b168a1be77","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 22:15:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8690080fefdf7f91745798df53595cb5a61fa592","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 22:25:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec88b5e369ed2871b5473c8bae27509d0caa3cf9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 22:28:24.000000000","message":"Topic set to bind-to-mgmt","accounts_in_message":[],"_revision_number":2},{"id":"fb2523517462f825daf4e51085bce26972b71d8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 00:02:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/03/670403/2/check/openstack-ansible-linters/2b22169/ : SUCCESS in 6m 12s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/03/670403/2/check/openstack-ansible-functional-debian-stable/5f97062/ : SUCCESS in 16m 09s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/03/670403/2/check/openstack-ansible-functional-centos-7/125b0b0/ : SUCCESS in 19m 51s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/03/670403/2/check/openstack-ansible-functional-ubuntu-bionic/40f96c3/ : SUCCESS in 16m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/670403/2/check/openstack-ansible-deploy-aio_metal-debian-stable/8aa3c9a/ : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/670403/2/check/openstack-ansible-deploy-aio_metal-centos-7/61c7442/ : SUCCESS in 1h 30m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/03/670403/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/17d52e7/ : SUCCESS in 1h 23m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/670403/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ce9fb3e/ : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/03/670403/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e69ab95/ : SUCCESS in 58m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/03/670403/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a2658af/ : FAILURE in 21m 37s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/03/670403/2/check/openstack-tox-docs/a4433b5/html/ : SUCCESS in 3m 56s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/03/670403/2/check/openstack-ansible-functional-centos-7-ssl/26f28cf/ : SUCCESS in 26m 57s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/03/670403/2/check/openstack-ansible-functional-opensuse-15-ssl/b24e4c5/ : FAILURE in 40m 10s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/03/670403/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/41711c5/ : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ca79d872c5726c590e7fd19469feb93414a72b22","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-12 08:44:48.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"636e30fc4f77d2a9f9d2649d7c6167a175c273bc","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-11 15:13:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3aa123338750d84e4654dc3b149738fc5f07047f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 17:11:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0005df786ed341458e28dee575aa43c1 : SUCCESS in 7m 27s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/c6569df58b8442ad93cbfeeaf4ff0c32 : SUCCESS in 15m 31s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/c3df5d3e6c9a4b6b87607f0911293f9f : SUCCESS in 20m 12s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4f73c42234c4fd7a1daadfe6ca4d093 : SUCCESS in 26m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ba64b9f16dd04e62bd9697377c472420 : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92871a8678ef4962951569a1ec0e56f2 : SUCCESS in 1h 42m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7b667ff1256a471eb62423c8596f4ff6 : SUCCESS in 1h 23m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42cbccec45744997a99f945b0c1054ad : SUCCESS in 1h 09m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d40923ac5c714e54861c8fd9c3642804 : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c7c2abc8fdf048e6bb10c92ed82e2a58 : FAILURE in 37m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/692b76a3048e4ce091853c71eef1794b : SUCCESS in 1h 02m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b93a4bfe7d1b48d7a466d81516ab18b3 : SUCCESS in 3m 21s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/440258559f58499d848975845d0d55d7 : SUCCESS in 22m 57s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/86723a0ce6f34c38adf6ff2cac54ca09 : FAILURE in 34m 14s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/c10d451992824d049f983c2bf2d78484 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":3},{"id":"2a00ee3eb8f8b411d61df8169629b03381314ba3","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-11 18:56:05.000000000","message":"Abandoned\n\nlooking to some multi node metal deployments, we don\u0027t need these variables but just some overrides in group vars as it is being done here https://review.opendev.org/#/c/670051/21","accounts_in_message":[],"_revision_number":3},{"id":"39edfba861559717a18e1c1c86f0256fdcaec2ad","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-03 19:55:02.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"b98b05bf0007143bfd6f2bf497cc98f3c063d65c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 20:03:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"273f4a024da9197170e2d7b2ca1814f51df5f4a5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 20:06:55.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c0ebc0862361edcb04c7b865c971ac6841104e69","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 20:07:02.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"f84d6a9b4854b7f79baa86af5d3514181558746d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-03 20:47:58.000000000","message":"Patch Set 6:\n\nI think this is going to need login_unix_socket: \"{{ galera_unix_socket }}\" on the mysql_user task in galera_server_setup.yml","accounts_in_message":[],"_revision_number":6},{"id":"38e49eecf8b1111cfceaeb83169b3983e1f76da3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-03 20:48:37.000000000","message":"Patch Set 6:\n\n\u003e I think this is going to need login_unix_socket: \"{{\n \u003e galera_unix_socket }}\" on the mysql_user task in galera_server_setup.yml\n\nAh this is rebased on the 10.4 patch, that should be in place","accounts_in_message":[],"_revision_number":6},{"id":"99c784f60105dcd3f9ec42f6619d7e8f21999c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 22:06:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9d58b90afd8844df8ab03b032ee24cda : SUCCESS in 7m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/668dc9ceac584f39a374331cd6911050 : SUCCESS in 22m 28s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/2ba09662885145caaebcea27e89451d9 : FAILURE in 20m 55s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/1dd7b2d7d6eb4d298b442d26521fb827 : FAILURE in 7m 09s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e13e5eefbf38447d8188252fe51ad999 : SUCCESS in 20m 56s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fffebc72300f47a29c8a7589df52236c : FAILURE in 7m 39s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c895a7c72ac64f089dd15dc228c4dd90 : SUCCESS in 52m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6f95b777fb43498cbfc966da8c6d13c1 : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db4b0c94f9d24c2081208988dcef109d : SUCCESS in 1h 08m 21s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ff8d64de14fe4c879265eca1dbf3b721 : SUCCESS in 1h 33m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65de3b3bb937463e860d8fceff5c7067 : SUCCESS in 1h 58m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cf7b0d8253334a188fe082cba9981779 : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d70c689639446dad69e007c2d278e0 : SUCCESS in 44m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967ab23a4ba64db8a67315b4d653d0c2 : SUCCESS in 5m 38s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/8e0a3605eea3460c91f16f120ce8f4e5 : FAILURE in 37m 15s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/410bcf063b414054862306eb8a400312 : FAILURE in 16m 17s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/58b1d81946ab4e2eb61d994955a855e4 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":6},{"id":"00d1635e07769145415a06a507873c4b6be15866","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-04 19:32:49.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2a685bc43c70ebc55c0b7f4866c920a9196af58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 22:19:13.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3ae54ab2605c491b83ec8414b4f0af59 : SUCCESS in 7m 29s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/2c842b0647864b18ae95640972a62f1b : SUCCESS in 18m 22s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/7919be6d649b47e69c8a9649d27a6d1f : SUCCESS in 24m 01s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/139d26414bbc4b049f6a4f17b055d06f : FAILURE in 10m 30s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e3ace586cf24d6aa22612d1a91ad83b : SUCCESS in 19m 27s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc0273db79b24755ad46c0d4d7ae728d : FAILURE in 9m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/60caf5e5356b47d5b72cf42a54b3be57 : SUCCESS in 1h 07m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eaf47c022e114ac7a64f1b2dfeaabbb1 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f909176713fe4d94ad2656c5db122fae : SUCCESS in 1h 08m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/163e8414c49943468d5f6b5520fe2c81 : SUCCESS in 2h 39m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59ed3cb51e034203b9de33424c8c5613 : SUCCESS in 1h 54m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b02ac333f27349bfad3fcee08decbf01 : SUCCESS in 1h 11m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74059072b54b44fb8dd801cf197bd1d8 : SUCCESS in 55m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39628e2c49484dd3896da142247508a2 : SUCCESS in 9m 08s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/d0e1827efea041e8bbe3d01bbab33025 : SUCCESS in 21m 00s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/1a7f043d1e344dbbb1cb0aa7c96ca91c : FAILURE in 16m 15s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/51e6833260a0459a901667cb7a265120 : FAILURE in 32m 25s","accounts_in_message":[],"_revision_number":6},{"id":"b97ce4f1f44814cbbade0856c7fcbce4b617ddf6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-06 07:49:42.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"801ca7134bd77085614dea36ff46a14b1e488b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 10:32:45.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/110ad61347c945349ac3387513c82a22 : SUCCESS in 7m 03s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/631fbdc30f454f0284b9adaa78d1bf4b : SUCCESS in 19m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/6f72e72986544fe884542a5db2238df3 : SUCCESS in 18m 53s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/055025e5551649faafc5a26bc46caa35 : FAILURE in 8m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2049944277b943fa8dae1a9497b663eb : SUCCESS in 20m 05s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84db6af683bc457c974fa8cc8775eb4b : FAILURE in 5m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/563811a6184c45979bd43100a86e0a0d : FAILURE in 1h 29m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6444222187d244d89718906b52f2b5d7 : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/800bf4941fc04845bd1a64f3e2b037d7 : SUCCESS in 1h 10m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c5c9ac0da2074ac6bdf43cea705d5458 : SUCCESS in 2h 40m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/197be2c5222744debeb51c7b7e7ea197 : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5f59f3179cc4f5c9bd729753cbc06ad : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/667b8523134b4a61b1133baa95fba1b9 : SUCCESS in 1h 29m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124ce83affa3493392c3029b29b4b8b6 : SUCCESS in 5m 03s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/222b0fc69660461fb1c082d8e3332d3f : SUCCESS in 23m 21s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/a301127156ae47d7a651f73f51bba2a4 : FAILURE in 15m 37s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/5b7acf3cc1de4d849cd7d63ae260c356 : SUCCESS in 20m 49s","accounts_in_message":[],"_revision_number":6},{"id":"28a943a81119af2db75a0b6f449abe2db1ccc76e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-15 06:59:52.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ea7ae4ea1de8c3c53211735f2c70e5c62cb8573b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 08:48:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8b5b1457f46745ccb8f65ab2a5682016 : SUCCESS in 8m 36s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/d9d74f0891fa410dae134c0e908b9b7b : SUCCESS in 22m 04s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0d1285bc093b4e1d96b0cb6bc534edff : SUCCESS in 24m 30s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d8cf91473dc2490994f80d1a3cdda531 : FAILURE in 13m 44s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c3bacdf06b042e3a3feea1e24f55f91 : SUCCESS in 17m 59s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ba72532299443e4bfbaf184b286240e : FAILURE in 15m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4d8dd24620024f109d3d4b8f7adfe83f : SUCCESS in 54m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/42fa1ad75c0f478bbe7e60120000c7bb : RETRY_LIMIT in 7m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55132129a90343d4b21229a403a22283 : SUCCESS in 52m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e901ead863d347cba505ec304be3f205 : FAILURE in 1h 38m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed1bf5de1ff5480c9d8da4e68599838e : FAILURE in 1h 07m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8c61baf3d4eb43de8372ba1d49316e73 : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e0d50966d744aca2c4e06173558701 : SUCCESS in 53m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7157dc5b834515a6186033ad44562e : SUCCESS in 6m 47s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/f04c265c922f49118a231b8c74e08354 : SUCCESS in 21m 03s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/f03028f441814956bec75702fcfb60cf : FAILURE in 18m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/deeb7a6861d04f989272ad6b6a17760b : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":7},{"id":"ee2f6e1dc85ccf183388e617b75d1d84405cf85f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-15 10:22:40.000000000","message":"Patch Set 7: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0b854f50621d335ebc2499caab99ce5213e34ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 12:22:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5e87c988147a44a3933c72d7471f5fa0 : SUCCESS in 6m 16s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/490d5d4d391141d0a05f5d817af48bc5 : SUCCESS in 15m 46s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/366aaeaf04894a4eb6e9062e242784bc : SUCCESS in 25m 36s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/af4641e0bbb3487e8987aacba39a255e : FAILURE in 13m 22s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2539383bd63412591ade467bc5f0a9a : SUCCESS in 16m 51s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be40ba521bbc4a3eacf493df760c1e69 : FAILURE in 13m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4021e78d0b5a410da39923394448108b : SUCCESS in 53m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e24adca9bef4ccfbafe153c2a5f8f9f : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3e91049891d4318b52644c1d36a2912 : SUCCESS in 1h 08m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/73e4c7295a0c490ea1fb3aa5bedc8888 : FAILURE in 1h 42m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b21861239c88499abfc08f58c9943848 : SUCCESS in 1h 24m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bac3c7bc882b44b29f6e9486d81cc16e : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a34f05293a848989a0bf7dc76860488 : SUCCESS in 53m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38aad6d623f448081187af950572815 : SUCCESS in 5m 23s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/7939e1a6664b4d8193b4b80b74770bf8 : SUCCESS in 21m 44s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/f6dc23f12b784f61a3f74b1a015749a4 : FAILURE in 15m 17s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/1f67b8a77f6540d2a435152b9fcb556a : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":7},{"id":"3aed82167888dce09c1e34aa1c76062571a943d7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-15 19:36:02.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"05bedb7fff4941bf873851bb24f9702ad59d01a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:36:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"50d1ba95769149e6d654371e19a47538dec09c60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 21:49:59.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/485bd9853c804ea7af64adf6d1910123 : SUCCESS in 12m 07s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b4542f5df5a146a9aa8cb346ec899d49 : FAILURE in 31m 14s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/83b6665c49e34106a3e7af80743ce238 : RETRY_LIMIT in 5m 59s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f0e69133d7240b19f9ab9bd65028f71 : SUCCESS in 24m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7df9e50aa6064bbb946f679a7c9a88c9 : SUCCESS in 1h 15m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f7733af415945a6afa900514a12b142 : RETRY_LIMIT in 11m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dd697acf6374b02a5a8f154d8923c1e : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65ed1612acef4a3a968a3386cda4e567 : RETRY_LIMIT in 11m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da798ac9e2ba48c8ae8979b3762439d4 : SUCCESS in 1h 08m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06db1a92f764e44a50ff244a1fba5d2 : SUCCESS in 9m 21s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/1e5cab55351b4e628e82913e4396af26 : RETRY_LIMIT in 10m 16s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/fd842e94cf164f398055883cb3959fec : TIMED_OUT in 1h 34m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b76555e89d65404986446bb14f0da4e1 : SUCCESS in 2h 04m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2228ed9a72663077f50457fb89a7325a86b40d18","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-16 05:22:25.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"7ea0ae2131a570d814d44cd39380cab06aefb78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 07:46:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/78a1018991ab42b5b7e3b0c9718cca15 : SUCCESS in 6m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/61950f63883b4015ab51d1054b7f3e18 : SUCCESS in 24m 15s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/5d59d347460349b0adecbc4128450ab9 : SUCCESS in 19m 07s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d0e6f5f20e1d4757a6c4b306486dc9e7 : FAILURE in 9m 45s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/256f5396285941578a8b1e16f16a0042 : SUCCESS in 21m 00s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9f983b173f7441daa569243d9ca9bf0 : FAILURE in 14m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ccf8401b2b324781a8b60b1f76f894a0 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e251182ffdd548b19b1b75b723a827bb : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26208504b2aa4d748ddba889e14b3afd : SUCCESS in 1h 28m 07s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/adf5cd30427a40e984ddffb8c1bb7777 : FAILURE in 1h 35m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77d6511abe8e49b0b95f5e20a86f7790 : SUCCESS in 2h 21m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bbc155646cc84472941a94c860c67c2a : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3f0081577a64a45a617ce50aa962a19 : SUCCESS in 56m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae7eecf60184a68becedb2e9e5dedfa : SUCCESS in 5m 19s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/aca00ae35c8f40f992441180ab31c00b : SUCCESS in 25m 24s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/88c7d0bfb5f745ba865d57ee7551de64 : FAILURE in 13m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/9404fca6e7c042349298e8a12be7b12e : SUCCESS in 17m 41s","accounts_in_message":[],"_revision_number":8},{"id":"206540d5574e81f66aa990ee3cb7beb2f055fe00","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-16 07:55:40.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"5b6ee55e2a976ced619d1fc81923c795d9e284bf","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-16 14:25:40.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"9287e0f3d1307de38403227289be61e701c923c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 14:25:51.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"bf47efe5b6a619d9f6ef8661bf261e44b4b76232","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 18:09:43.000000000","message":"Patch Set 8:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dff0675b01f214a946fa81be67aaed3fab8f46ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:16:31.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8fc58f2b529e41808dcccac5a0df2818 : SUCCESS in 11m 51s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5d75ba2515934e66ab7ca8149586c7b6 : SUCCESS in 31m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b16de83e498442cf84aa6fe827559ed0 : SUCCESS in 48m 20s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/316a935fc0ff49ee8e0818bad1ab1d05 : SUCCESS in 24m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d36f426622384972922f833972d1eddf : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b8197cec409e4d70bcf71a6850c67803 : SUCCESS in 1h 03m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d4c534eac134fe4a1b15bdc36a7ba8f : SUCCESS in 56m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6d8a9fa49a07463f89f04c980a3128cd : SUCCESS in 1h 10m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26583e04719d4d68844ea5fc6faf11a2 : SUCCESS in 1h 05m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556d535a38124b9297bf61ab8a4a06bf : SUCCESS in 14m 30s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/5d5dd7b05aef4c5f8280978d09c491e0 : SUCCESS in 29m 58s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/e48efdd9f5484bbaab46231eca75915e : SUCCESS in 26m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b82e47a04cb44c43abcf7a425c124f6c : SUCCESS in 1h 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9a4a75bf1d81393857aa8ae8b3ab6e3c938e9f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:16:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"a80c6060025d56e534b477d2f86111761ee2b1a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:20:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0fd3d2dfaf4d2489f4b02253ddbf05 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":8},{"id":"d22ca7ee0ebb893f711d300a487f3c01b03d48d3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 13:32:31.000000000","message":"Patch Set 8: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 4735f98f391a9b1943a1320109ffbb5074e83d9c","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"955ee8da9bbdaaa5f898e00e324f657157ddbd74","revisions":{"766121f273732614903fb54346bd85a29cfeeb59":{"kind":"REWORK","_number":1,"created":"2019-07-11 22:15:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/670403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/1"}}},"commit":{"parents":[{"commit":"ea1f2688323009f845ae57fa7fe4a2bb1d7df674","subject":"Merge \"Replace opensuse 42.3 functional testing with opensuse-15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ea1f2688323009f845ae57fa7fe4a2bb1d7df674"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:14:22.000000000","tz":60},"subject":"Add defaults for wsrep cluster bind address and port","message":"Add defaults for wsrep cluster bind address and port\n\nThis is defaulted to 0.0.0.0:4567 and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/766121f273732614903fb54346bd85a29cfeeb59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/766121f273732614903fb54346bd85a29cfeeb59"}]},"branch":"refs/heads/master"},"587f4c3c1b9d179a933f8c39c8936e25a711252e":{"kind":"REWORK","_number":2,"created":"2019-07-11 22:25:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/670403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/2"}}},"commit":{"parents":[{"commit":"ea1f2688323009f845ae57fa7fe4a2bb1d7df674","subject":"Merge \"Replace opensuse 42.3 functional testing with opensuse-15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ea1f2688323009f845ae57fa7fe4a2bb1d7df674"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:24:13.000000000","tz":60},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/587f4c3c1b9d179a933f8c39c8936e25a711252e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/587f4c3c1b9d179a933f8c39c8936e25a711252e"}]},"branch":"refs/heads/master"},"0bce51f4d663a7c29924171858f8acb257bec771":{"kind":"REWORK","_number":3,"created":"2019-09-11 15:13:58.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/03/670403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/3"}}},"commit":{"parents":[{"commit":"ea1f2688323009f845ae57fa7fe4a2bb1d7df674","subject":"Merge \"Replace opensuse 42.3 functional testing with opensuse-15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ea1f2688323009f845ae57fa7fe4a2bb1d7df674"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-09-11 15:13:38.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0bce51f4d663a7c29924171858f8acb257bec771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0bce51f4d663a7c29924171858f8acb257bec771"}]},"branch":"refs/heads/master"},"dbdc8cb8bb7071c178a03b63014eea41ed38e4d4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-06-03 20:03:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/670403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/4"}}},"commit":{"parents":[{"commit":"70881b63255f3b79f654d136ca863e870b9ace74","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/70881b63255f3b79f654d136ca863e870b9ace74"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-03 20:03:32.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dbdc8cb8bb7071c178a03b63014eea41ed38e4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/dbdc8cb8bb7071c178a03b63014eea41ed38e4d4"}]},"branch":"refs/heads/master"},"7f4b246acb9b491be89886254df89d64eaa8e346":{"kind":"REWORK","_number":5,"created":"2020-06-03 20:06:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/670403/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/5"}}},"commit":{"parents":[{"commit":"70881b63255f3b79f654d136ca863e870b9ace74","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/70881b63255f3b79f654d136ca863e870b9ace74"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-03 20:05:54.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7f4b246acb9b491be89886254df89d64eaa8e346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7f4b246acb9b491be89886254df89d64eaa8e346"}]},"branch":"refs/heads/master"},"b6b06187ba303c45c65c53c64507f287e62ee67f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-03 20:07:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/670403/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/6"}}},"commit":{"parents":[{"commit":"0a36313fa0749c8862bdc8b327721e76200f9a88","subject":"Bump MariaDB version to 10.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/0a36313fa0749c8862bdc8b327721e76200f9a88"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-03 20:07:02.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b6b06187ba303c45c65c53c64507f287e62ee67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/b6b06187ba303c45c65c53c64507f287e62ee67f"}]},"branch":"refs/heads/master"},"5da80865c27a3752d0323625766f846e00f6cf07":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-06-15 06:59:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/670403/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/7"}}},"commit":{"parents":[{"commit":"9e5497aa0a01539e38cdf17e32b8c631e41dab06","subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e5497aa0a01539e38cdf17e32b8c631e41dab06"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-15 06:59:52.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5da80865c27a3752d0323625766f846e00f6cf07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5da80865c27a3752d0323625766f846e00f6cf07"}]},"branch":"refs/heads/master"},"955ee8da9bbdaaa5f898e00e324f657157ddbd74":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-06-16 05:22:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/670403/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/03/670403/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/03/670403/8"}}},"commit":{"parents":[{"commit":"9e5497aa0a01539e38cdf17e32b8c631e41dab06","subject":"Fix Mariadb 10.3 -\u003e 10.4 upgrade path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9e5497aa0a01539e38cdf17e32b8c631e41dab06"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 22:13:08.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-16 05:22:23.000000000","tz":0},"subject":"Add defaults for galera server and wrep cluster bind addresses","message":"Add defaults for galera server and wrep cluster bind addresses\n\nThis is defaulted and can be overridden in a deployment\nwhen the correct managment network address is known\n\nDepends-On: https://review.opendev.org/734672\nDepends-On: https://review.opendev.org/735289\nChange-Id: I4bb36e779002a6ce7a22fe308195de3b10aa1aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/955ee8da9bbdaaa5f898e00e324f657157ddbd74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/955ee8da9bbdaaa5f898e00e324f657157ddbd74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
