)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~670599","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I5805fb11a5118c4011701377cd30b2511edca505","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"I5805fb11a5118c4011701377cd30b2511edca505","subject":"cleanup: refactor into galera_mariadb_service_name","status":"MERGED","created":"2019-07-12 16:42:25.000000000","updated":"2019-07-16 15:43:52.000000000","submitted":"2019-07-16 15:43:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670599-1563291832965-fd40a304","meta_rev_id":"80007c86f89103fa630e96f50a03d0cb114f719a","_number":670599,"virtual_id_number":670599,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-07-16 15:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-07-16 14:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-15 21:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-07-16 14:01:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 21:04:18.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-16 14:01:04.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-16 15:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8d17cb937dbc8a5af7c312abcb56b492783e2f4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 16:42:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"406ec80077275726f5be47ab54875bbe169af90a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 18:29:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/670599/1/check/openstack-ansible-linters/9a56ca2/ : SUCCESS in 7m 45s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/99/670599/1/check/openstack-ansible-functional-debian-stable/3be611f/ : SUCCESS in 17m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/670599/1/check/openstack-ansible-functional-centos-7/df59cdb/ : SUCCESS in 19m 45s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/670599/1/check/openstack-ansible-functional-ubuntu-bionic/10bfc5d/ : SUCCESS in 18m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/670599/1/check/openstack-ansible-deploy-aio_metal-debian-stable/a5c8a45/ : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/670599/1/check/openstack-ansible-deploy-aio_metal-centos-7/460751d/ : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/670599/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/a4d713b/ : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/670599/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5cac75f/ : SUCCESS in 1h 24m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/670599/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/fb8be28/ : SUCCESS in 1h 03m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/99/670599/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/01cddae/ : SUCCESS in 54m 11s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/99/670599/1/check/openstack-tox-docs/ca5d34d/html/ : SUCCESS in 4m 07s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/99/670599/1/check/openstack-ansible-functional-centos-7-ssl/3135012/ : SUCCESS in 23m 48s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/99/670599/1/check/openstack-ansible-functional-opensuse-15-ssl/c5b9a6c/ : FAILURE in 15m 28s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/99/670599/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/080b738/ : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"047a3f000798ea093bd3d1daf94a367478e11f8d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-15 21:04:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc4533b8c47dd68235379e2c74e1983300b65040","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-16 14:01:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9806c6b984d164d2324a6da5d2fba7de37727a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:01:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c237fad1fb7dc37c568e5be8c67546a9dc448c05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:43:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/99/670599/1/gate/openstack-ansible-linters/5296b1b/ : SUCCESS in 6m 18s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/99/670599/1/gate/openstack-ansible-functional-debian-stable/6dfe653/ : SUCCESS in 14m 22s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/99/670599/1/gate/openstack-ansible-functional-centos-7/81d6350/ : SUCCESS in 18m 47s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/99/670599/1/gate/openstack-ansible-functional-ubuntu-bionic/11c88ee/ : SUCCESS in 17m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/670599/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/2eb6229/ : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/670599/1/gate/openstack-ansible-deploy-aio_metal-centos-7/4cc8faf/ : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/670599/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/621357e/ : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/670599/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/663dab0/ : SUCCESS in 1h 13m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/670599/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/0d0eab3/ : SUCCESS in 1h 15m 05s\n- openstack-tox-docs http://logs.openstack.org/99/670599/1/gate/openstack-tox-docs/3a0aecb/html/ : SUCCESS in 5m 07s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/99/670599/1/gate/openstack-ansible-functional-centos-7-ssl/0f54a6a/ : SUCCESS in 18m 11s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/99/670599/1/gate/openstack-ansible-functional-ubuntu-bionic-ssl/cccc140/ : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"80007c86f89103fa630e96f50a03d0cb114f719a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:43:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c7ade4fc90b12f73099752be486d35b6cc0dae4","revisions":{"7c7ade4fc90b12f73099752be486d35b6cc0dae4":{"kind":"REWORK","_number":1,"created":"2019-07-12 16:42:25.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/99/670599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/99/670599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/99/670599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/99/670599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/99/670599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/99/670599/1"}}},"commit":{"parents":[{"commit":"ea1f2688323009f845ae57fa7fe4a2bb1d7df674","subject":"Merge \"Replace opensuse 42.3 functional testing with opensuse-15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ea1f2688323009f845ae57fa7fe4a2bb1d7df674"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-12 16:41:41.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-12 16:42:24.000000000","tz":-240},"subject":"cleanup: refactor into galera_mariadb_service_name","message":"cleanup: refactor into galera_mariadb_service_name\n\nWe had two variable names which were duplicated for the service\nname, this converges them all to \"mariadb\" which uses the actual\nsystemd unit instead of the init.d compatiblity layer.\n\nChange-Id: I5805fb11a5118c4011701377cd30b2511edca505\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c7ade4fc90b12f73099752be486d35b6cc0dae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c7ade4fc90b12f73099752be486d35b6cc0dae4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
