)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~670601","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Ife23cf46ab06a81776f3e8653a40bd4bbf745b50","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Ife23cf46ab06a81776f3e8653a40bd4bbf745b50","subject":"setup: don\u0027t try to login to create users and upgrade","status":"MERGED","created":"2019-07-12 16:52:52.000000000","updated":"2019-07-18 20:26:59.000000000","submitted":"2019-07-16 23:21:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670601-1563319311327-8d65fae4","meta_rev_id":"3d03fe054ed25e90809fe015c53387422e3d30f2","_number":670601,"virtual_id_number":670601,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-07-16 23:21:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-07-12 19:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-12 19:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2019-07-12 19:30:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 19:24:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-12 19:30:39.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-16 23:21:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2307d4fe74480c65977d6bafccf5bbe171fc1c1a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 16:52:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e5c5c53e9afd3821da847976172d4e9bfbcfffa","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-12 16:54:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9fa878812c70c94d06cfc859b95a927017ec434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 18:28:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/01/670601/2/check/openstack-ansible-linters/e89eb5c/ : SUCCESS in 6m 54s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-debian-stable/f6e7096/ : SUCCESS in 15m 25s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-centos-7/fc74f7c/ : SUCCESS in 22m 10s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-ubuntu-bionic/b00c17a/ : SUCCESS in 14m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-debian-stable/b3c5831/ : SUCCESS in 1h 05m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-centos-7/fd80d7f/ : SUCCESS in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/cec6c36/ : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9669891/ : SUCCESS in 59m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/048b7cc/ : SUCCESS in 55m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/96f4a15/ : SUCCESS in 1h 11m 57s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/01/670601/2/check/openstack-tox-docs/9e2572c/html/ : SUCCESS in 3m 23s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-centos-7-ssl/acc023f/ : SUCCESS in 22m 57s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-opensuse-15-ssl/8d862ed/ : FAILURE in 18m 05s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/a5af364/ : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7558cfae6ccdcfc06f53d04db092169f3f2c689d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-12 19:24:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"752030af5984562f10f3de2a977daab7bade0e0a","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-12 19:30:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0129c6afb1ea0d1eb978f9170a814a93327a51c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:01:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38ca0316a64ca8c2c6947e7d920673624c734967","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:43:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/01/670601/2/gate/openstack-ansible-linters/7371b01/ : SUCCESS in 8m 31s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-debian-stable/a622502/ : SUCCESS in 16m 18s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-centos-7/7fd7ab3/ : SUCCESS in 19m 42s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-ubuntu-bionic/165bc8c/ : SUCCESS in 20m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/1c68e66/ : FAILURE in 14m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-centos-7/44db2f0/ : SUCCESS in 1h 10m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/a5eb52c/ : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8dc54b9/ : SUCCESS in 1h 29m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/4c15fd5/ : SUCCESS in 1h 17m 28s\n- openstack-tox-docs http://logs.openstack.org/01/670601/2/gate/openstack-tox-docs/26eca49/html/ : SUCCESS in 3m 11s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-centos-7-ssl/7fe982f/ : SUCCESS in 18m 17s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-ubuntu-bionic-ssl/dbc63f8/ : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"cfc961ca615cc98ae7a77ded41229d674cdb4233","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-16 20:01:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e76e0d45351a8d83351735ec5863a21bc1e5a0c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:40:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/01/670601/2/check/openstack-ansible-linters/0ce3484/ : SUCCESS in 7m 06s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-debian-stable/369435c/ : SUCCESS in 17m 25s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-centos-7/ad770ee/ : SUCCESS in 18m 55s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-ubuntu-bionic/b1e9973/ : SUCCESS in 15m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-debian-stable/42a2a9b/ : SUCCESS in 1h 06m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-centos-7/9651607/ : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/a59112b/ : SUCCESS in 1h 09m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d1f3212/ : SUCCESS in 1h 16m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/fcb16f2/ : SUCCESS in 1h 01m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/01/670601/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/bdb96e4/ : SUCCESS in 1h 10m 26s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/01/670601/2/check/openstack-tox-docs/b40a18f/html/ : SUCCESS in 3m 07s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-centos-7-ssl/0ccf111/ : SUCCESS in 22m 21s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-opensuse-15-ssl/3309bc9/ : FAILURE in 37m 51s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/01/670601/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/efc2a23/ : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":2},{"id":"41996ba49cc013be6778e8ae07840a3676a4e2ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:41:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c99f27254157366a6ad2bc033d353845d62f849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:21:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d7cab0548a3831d265c9f93385364192095101a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:21:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/01/670601/2/gate/openstack-ansible-linters/9061c47/ : SUCCESS in 5m 44s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-debian-stable/2c7b09c/ : SUCCESS in 15m 50s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-centos-7/0de1de2/ : SUCCESS in 17m 56s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-ubuntu-bionic/e760640/ : SUCCESS in 16m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/250c968/ : SUCCESS in 1h 01m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-centos-7/dcc782b/ : SUCCESS in 1h 31m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/c96bda9/ : SUCCESS in 1h 28m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/20da1e5/ : SUCCESS in 1h 13m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/01/670601/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/a917a41/ : SUCCESS in 1h 07m 15s\n- openstack-tox-docs http://logs.openstack.org/01/670601/2/gate/openstack-tox-docs/4ff78f6/html/ : SUCCESS in 2m 45s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-centos-7-ssl/4f0ab42/ : SUCCESS in 19m 58s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/01/670601/2/gate/openstack-ansible-functional-ubuntu-bionic-ssl/9318bf7/ : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3d03fe054ed25e90809fe015c53387422e3d30f2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 20:26:59.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit cfe235ecc4c89143f4e99e36f4714cf6842026a5","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f73e4a1c1527c1593cdc98cf13b39cc60dff080","revisions":{"598311c358444dade8ba0f4ef703e85d604e7fea":{"kind":"REWORK","_number":1,"created":"2019-07-12 16:52:52.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/01/670601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/670601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/1"}}},"commit":{"parents":[{"commit":"7c7ade4fc90b12f73099752be486d35b6cc0dae4","subject":"cleanup: refactor into galera_mariadb_service_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c7ade4fc90b12f73099752be486d35b6cc0dae4"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-12 16:51:29.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-12 16:52:05.000000000","tz":-240},"subject":"setup: don\u0027t try to login to create users","message":"setup: don\u0027t try to login to create users\n\nWe delegate to a setup host which already has things wired up\nfor setting things up which contains a .my.cnf file that works\njust fine with all credentials to connect.\n\nChange-Id: Ife23cf46ab06a81776f3e8653a40bd4bbf745b50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/598311c358444dade8ba0f4ef703e85d604e7fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/598311c358444dade8ba0f4ef703e85d604e7fea"}]},"branch":"refs/heads/master"},"9f73e4a1c1527c1593cdc98cf13b39cc60dff080":{"kind":"REWORK","_number":2,"created":"2019-07-12 16:54:51.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/01/670601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/01/670601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/01/670601/2"}}},"commit":{"parents":[{"commit":"7c7ade4fc90b12f73099752be486d35b6cc0dae4","subject":"cleanup: refactor into galera_mariadb_service_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7c7ade4fc90b12f73099752be486d35b6cc0dae4"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-12 16:51:29.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-07-12 16:54:20.000000000","tz":-240},"subject":"setup: don\u0027t try to login to create users and upgrade","message":"setup: don\u0027t try to login to create users and upgrade\n\nWe delegate to a setup host which already has things wired up\nfor setting things up which contains a .my.cnf file that works\njust fine with all credentials to connect.\n\nChange-Id: Ife23cf46ab06a81776f3e8653a40bd4bbf745b50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9f73e4a1c1527c1593cdc98cf13b39cc60dff080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/9f73e4a1c1527c1593cdc98cf13b39cc60dff080"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
