)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~672059","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I6add49391e7544c40375c12ea803db08838385af","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"I6add49391e7544c40375c12ea803db08838385af","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-22 14:32:58.000000000","updated":"2019-07-29 22:28:47.000000000","submitted":"2019-07-29 22:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672059-1564439327536-f6dc5ada","meta_rev_id":"c3bc37ee710d2205cfbde6176a746e73335b8428","_number":672059,"virtual_id_number":672059,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-29 22:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-29 20:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-27 21:21:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"date":"2019-07-23 14:02:02.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-29 20:47:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-27 21:21:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-29 20:47:06.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-29 22:28:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1b615486707339254e64457ed58cbd3a1e76755","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-22 14:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc53124c3de750d8732b09fbc1ec6423f09688d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 16:12:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/672059/1/check/openstack-ansible-linters/570eabc/ : SUCCESS in 5m 46s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/59/672059/1/check/openstack-ansible-functional-debian-stable/b39682d/ : SUCCESS in 16m 21s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/672059/1/check/openstack-ansible-functional-centos-7/881a8f0/ : SUCCESS in 21m 04s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/672059/1/check/openstack-ansible-functional-ubuntu-bionic/6f74bf6/ : SUCCESS in 17m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_metal-debian-stable/6edd519/ : SUCCESS in 1h 10m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_metal-centos-7/11cc46f/ : SUCCESS in 1h 34m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/d3e04a7/ : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1da793a/ : SUCCESS in 1h 29m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5b65c2c/ : SUCCESS in 1h 03m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/4c1d7e5/ : FAILURE in 50m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/59/672059/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/281b4c0/ : SUCCESS in 48m 31s\n- openstack-tox-docs http://logs.openstack.org/59/672059/1/check/openstack-tox-docs/f53026e/html/ : SUCCESS in 3m 41s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/59/672059/1/check/openstack-ansible-functional-centos-7-ssl/a190193/ : SUCCESS in 22m 59s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/59/672059/1/check/openstack-ansible-functional-opensuse-15-ssl/8a9b3c9/ : FAILURE in 27m 30s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/59/672059/1/check/openstack-ansible-functional-ubuntu-bionic-ssl/9ad9e0a/ : SUCCESS in 16m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7ed980a2c0aeb8d78d0052611c6c28340d08e849","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-22 16:32:30.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3f66d74056aeb99bb47a70a4bd8701c681bd9810","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-22 16:40:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9cca0387a4465805111929e09ec5df125fd3e02c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-22 16:48:10.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c91b1e50ca8c63164e691dde0e54666275eb1128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:07:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/672059/2/check/openstack-ansible-linters/d7c4156/ : SUCCESS in 6m 02s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/59/672059/2/check/openstack-ansible-functional-debian-stable/1fba403/ : SUCCESS in 22m 38s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/672059/2/check/openstack-ansible-functional-centos-7/57aed34/ : SUCCESS in 24m 39s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/672059/2/check/openstack-ansible-functional-ubuntu-bionic/8941796/ : SUCCESS in 23m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_metal-debian-stable/710ad8d/ : SUCCESS in 1h 05m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_metal-centos-7/9fc23cf/ : SUCCESS in 1h 19m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/93a45bb/ : SUCCESS in 1h 16m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7cd4f98/ : SUCCESS in 1h 08m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4980fac/ : SUCCESS in 1h 03m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/0fd256c/ : FAILURE in 56m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/59/672059/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/95afd69/ : SUCCESS in 51m 47s\n- openstack-tox-docs http://logs.openstack.org/59/672059/2/check/openstack-tox-docs/ea410c7/html/ : SUCCESS in 3m 52s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/59/672059/2/check/openstack-ansible-functional-centos-7-ssl/76a5f64/ : SUCCESS in 27m 50s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/59/672059/2/check/openstack-ansible-functional-opensuse-15-ssl/0d853b9/ : FAILURE in 32m 56s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/59/672059/2/check/openstack-ansible-functional-ubuntu-bionic-ssl/9c0a484/ : SUCCESS in 24m 16s","accounts_in_message":[],"_revision_number":2},{"id":"67c0d66fef6cac1bd6f14814600ba862d4504df2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 11:02:44.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"78e61782f028f8a7f1875f0b11bf297b9056c30c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 11:02:48.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"66a0e743799da3de4d2bc3cdbde61800ffd4702f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:48:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/59/672059/3/check/openstack-ansible-linters/7db9b35/ : SUCCESS in 4m 41s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-debian-stable/83975b4/ : SUCCESS in 16m 42s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-centos-7/d0054a4/ : FAILURE in 7m 51s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-ubuntu-bionic/8790a5e/ : SUCCESS in 17m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-debian-stable/655284c/ : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-centos-7/f69464e/ : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/fa8e5d6/ : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b2d6e06/ : SUCCESS in 1h 03m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/d9e84c0/ : SUCCESS in 57m 57s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/29a9fce/ : FAILURE in 56m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/1eb44e5/ : SUCCESS in 1h 02m 45s\n- openstack-tox-docs http://logs.openstack.org/59/672059/3/check/openstack-tox-docs/b8555b6/html/ : SUCCESS in 4m 31s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-centos-7-ssl/b96b7eb/ : SUCCESS in 25m 48s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-opensuse-15-ssl/84a2150/ : FAILURE in 31m 43s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-ubuntu-bionic-ssl/d1c75fd/ : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":3},{"id":"1e786f8fb15e0182bbc5c6663c3331c006a34c91","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 14:01:14.000000000","message":"Removed Workflow-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"dfe69df228255a66ad50599355d165bd514e34b2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-23 14:02:02.000000000","message":"Patch Set 3:\n\nrecheck\n\nThis is good to go, since journald collection will appear with patch to openstack-ansible\nhttp://logs.openstack.org/83/671783/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7902cfd/logs/host/mariadb.service.journal-12-31-37.log.txt.gz","accounts_in_message":[],"_revision_number":3},{"id":"0c0d23186bfffa3308c4e1686cafbb407503e4b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:55:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/672059/3/check/openstack-ansible-linters/cf2f08a/ : SUCCESS in 7m 19s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-debian-stable/09b1d30/ : SUCCESS in 17m 56s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-centos-7/7ede179/ : SUCCESS in 23m 05s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-ubuntu-bionic/c9a6b2c/ : SUCCESS in 22m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-debian-stable/2c8ccc1/ : SUCCESS in 1h 11m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-centos-7/944536e/ : SUCCESS in 1h 48m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/4c9c8f8/ : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ee44c92/ : SUCCESS in 57m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3ed1d4a/ : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/d5bc30f/ : FAILURE in 46m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/59/672059/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/cdfc7d6/ : SUCCESS in 47m 01s\n- openstack-tox-docs http://logs.openstack.org/59/672059/3/check/openstack-tox-docs/99e5246/html/ : SUCCESS in 3m 02s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-centos-7-ssl/86485a1/ : SUCCESS in 22m 14s\n- openstack-ansible-functional-opensuse-15-ssl http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-opensuse-15-ssl/2c8a1e4/ : FAILURE in 27m 14s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/59/672059/3/check/openstack-ansible-functional-ubuntu-bionic-ssl/e413bcd/ : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":3},{"id":"8bf12ccd4cd5e237618291258e43b598a87a623b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-27 21:21:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b336403f4b9194a1675bebc71a8c1c0063726489","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-29 20:47:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bd43eea60aabaed26c841a41b8a614aa453c9482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 20:47:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c066e99c9e865d0299e47d3db4bfa92881c6902a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:28:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/59/672059/3/gate/openstack-ansible-linters/5244943/ : SUCCESS in 6m 47s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/59/672059/3/gate/openstack-ansible-functional-debian-stable/3e8a51c/ : SUCCESS in 15m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/59/672059/3/gate/openstack-ansible-functional-centos-7/f7e8ef0/ : SUCCESS in 20m 18s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/59/672059/3/gate/openstack-ansible-functional-ubuntu-bionic/cb3227d/ : SUCCESS in 20m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/59/672059/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/1bd618a/ : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/59/672059/3/gate/openstack-ansible-deploy-aio_metal-centos-7/6b7a681/ : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/59/672059/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/79830d8/ : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/59/672059/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a4dc2a2/ : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/59/672059/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/981c0fe/ : SUCCESS in 1h 08m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/59/672059/3/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/f9c3840/ : SUCCESS in 52m 46s\n- openstack-tox-docs http://logs.openstack.org/59/672059/3/gate/openstack-tox-docs/c694309/html/ : SUCCESS in 4m 04s\n- openstack-ansible-functional-centos-7-ssl http://logs.openstack.org/59/672059/3/gate/openstack-ansible-functional-centos-7-ssl/85e130a/ : SUCCESS in 21m 24s\n- openstack-ansible-functional-ubuntu-bionic-ssl http://logs.openstack.org/59/672059/3/gate/openstack-ansible-functional-ubuntu-bionic-ssl/4f2340e/ : SUCCESS in 24m 23s","accounts_in_message":[],"_revision_number":3},{"id":"1b2a1bf1899b14d5aefe36019dcc9f7357e4feaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 22:28:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f935952931263ca4766549b1b0b9e4227dd48ebd","revisions":{"55f9dd5ee98310b31ac8d00ec8f681e40d5511b3":{"kind":"REWORK","_number":1,"created":"2019-07-22 14:32:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/672059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/59/672059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/1"}}},"commit":{"parents":[{"commit":"1d058a083b5d3f7b5c560bb1fde9829849bdf0a1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d058a083b5d3f7b5c560bb1fde9829849bdf0a1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-22 14:32:30.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-22 14:32:53.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: I6add49391e7544c40375c12ea803db08838385af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/55f9dd5ee98310b31ac8d00ec8f681e40d5511b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/55f9dd5ee98310b31ac8d00ec8f681e40d5511b3"}]},"branch":"refs/heads/master"},"549a176c79b6f1e5152e6d20c51ded02007cc58a":{"kind":"REWORK","_number":2,"created":"2019-07-22 16:40:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/672059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/59/672059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/2"}}},"commit":{"parents":[{"commit":"1d058a083b5d3f7b5c560bb1fde9829849bdf0a1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d058a083b5d3f7b5c560bb1fde9829849bdf0a1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-22 14:32:30.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-22 16:40:28.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: I6add49391e7544c40375c12ea803db08838385af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/549a176c79b6f1e5152e6d20c51ded02007cc58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/549a176c79b6f1e5152e6d20c51ded02007cc58a"}]},"branch":"refs/heads/master"},"f935952931263ca4766549b1b0b9e4227dd48ebd":{"kind":"REWORK","_number":3,"created":"2019-07-23 11:02:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/672059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/59/672059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/59/672059/3"}}},"commit":{"parents":[{"commit":"1d058a083b5d3f7b5c560bb1fde9829849bdf0a1","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/1d058a083b5d3f7b5c560bb1fde9829849bdf0a1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-22 14:32:30.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-23 11:02:37.000000000","tz":0},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald.\n\nChange-Id: I6add49391e7544c40375c12ea803db08838385af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f935952931263ca4766549b1b0b9e4227dd48ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f935952931263ca4766549b1b0b9e4227dd48ebd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
