)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~673624","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I5b5e514f8ffb2037e3a88b5b23f16e33f8062d64","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I5b5e514f8ffb2037e3a88b5b23f16e33f8062d64","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-30 19:34:47.000000000","updated":"2019-07-30 20:11:21.000000000","submitted":"2019-07-30 20:11:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673624-1564517481537-8da850ff","meta_rev_id":"5d98ce9fffdd95dc8c41f3b0b7348276d8384f1f","_number":673624,"virtual_id_number":673624,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 20:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-30 19:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-30 19:54:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 19:54:49.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-30 20:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51433f882b5d033ebdce55b291611d609e7cc686","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 19:34:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9a87192a2ce2010a0930fb5f8e29c22770f94db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:47:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/673624/1/check/openstack-ansible-linters/546f94e/ : SUCCESS in 5m 14s\n- requirements-check http://logs.openstack.org/24/673624/1/check/requirements-check/f12d037/ : SUCCESS in 3m 09s\n- openstack-tox-docs http://logs.openstack.org/24/673624/1/check/openstack-tox-docs/1f80f1a/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/24/673624/1/check/build-openstack-releasenotes/0fc72ae/html/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8323e982239ed782828ddb88e6c974a5f3f4beae","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 19:54:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22b4da6f9bcab2c8ccc51fade97aa5880c166b78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:55:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0c2cbcde425370c45f5da8ab81a8d5927741293","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:11:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c722910d10767c9e8b9c13eedd00c763e9bc2bc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:11:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/673624/1/gate/openstack-ansible-linters/2e15702/ : SUCCESS in 9m 40s\n- requirements-check http://logs.openstack.org/24/673624/1/gate/requirements-check/36aeefd/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/24/673624/1/gate/openstack-tox-docs/0fb2135/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/24/673624/1/gate/build-openstack-releasenotes/e644bb7/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f1115cee968e6b93ca80d2b7a127d5c75857278","revisions":{"5f1115cee968e6b93ca80d2b7a127d5c75857278":{"kind":"REWORK","_number":1,"created":"2019-07-30 19:34:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/673624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/24/673624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/673624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/673624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/673624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/24/673624/1"}}},"commit":{"parents":[{"commit":"f935952931263ca4766549b1b0b9e4227dd48ebd","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f935952931263ca4766549b1b0b9e4227dd48ebd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:34:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:34:43.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I5b5e514f8ffb2037e3a88b5b23f16e33f8062d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5f1115cee968e6b93ca80d2b7a127d5c75857278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/5f1115cee968e6b93ca80d2b7a127d5c75857278"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
