)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~677132","triplet_id":"openstack%2Fopenstack-ansible-galera_server~stable%2Frocky~I639f869084e71575b6db7a69fe9d06f67b25cfc9","project":"openstack/openstack-ansible-galera_server","branch":"stable/rocky","topic":"goodbye_42.3","hashtags":[],"change_id":"I639f869084e71575b6db7a69fe9d06f67b25cfc9","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-08-19 07:46:28.000000000","updated":"2019-08-27 13:31:23.000000000","submitted":"2019-08-27 13:31:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677132-1566912683988-1d226e13","meta_rev_id":"e4102775020f0ca5f80344f43ec0b6058afae31a","_number":677132,"virtual_id_number":677132,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-26 20:01:31.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-27 13:31:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-08-27 12:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-08-27 13:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-08-27 12:36:02.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-08-27 13:01:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 17:18:42.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-27 12:36:02.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-27 13:01:34.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-27 13:31:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bee4487ea171eaad556d392e278d0608830cbb4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 07:46:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"516bfd95272020d3da9e915bdabd49f4114bcbc2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 07:46:40.000000000","message":"Topic changed from goodbye_42.3-stable/rocky to goodbye_42.3","accounts_in_message":[],"_revision_number":1},{"id":"b3ee176c23a5252c97bc1619e15cda1d59628a18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 08:11:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/95eba80fbbbb44ed92cb6970b9672287 : SUCCESS in 5m 29s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1164003bd25b426b92f2b4c4899b9989 : POST_FAILURE in 19m 46s\n- openstack-ansible-functional-opensuse-423 https://zuul.opendev.org/t/openstack/build/709a13b9afb942d8aab7602f221be1aa : POST_FAILURE in 6m 10s (non-voting)\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/f8d3a94d7010482587d0dd793905d741 : POST_FAILURE in 8m 28s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8779d9db29114af898900e01fa43e858 : SUCCESS in 18m 55s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/010e413c0488485f808982619917071c : SUCCESS in 16m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/df685835cec54f758cb278b357a6b653 : SUCCESS in 3m 35s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ec7c55f9b739456092e5042a21cc16a0 : POST_FAILURE in 18m 30s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/ec90f26d482140f2b537aa5f46a52313 : SUCCESS in 16m 24s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a1ceaffee13b4d5a8de2373d2c7240d1 : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e496ca9f1b01a666cd357402f7af6198c499a950","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-19 17:18:42.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck -- code is good.","accounts_in_message":[],"_revision_number":1},{"id":"0270f791e5129524a13182d301c450a90c4032e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 18:13:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ff8e903984624f738f97278a0563e73f : SUCCESS in 7m 48s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/668d880503de4141a65bf9e211a3912d : POST_FAILURE in 21m 42s\n- openstack-ansible-functional-opensuse-423 https://zuul.opendev.org/t/openstack/build/7c6a1d0b222a472a8f20a7274bd77b53 : POST_FAILURE in 9m 42s (non-voting)\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/1287dfd817a147ea957399245c6d4660 : POST_FAILURE in 11m 12s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34f6469bae0e47fdb64250e7844fc156 : SUCCESS in 21m 07s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/06892ee3e41147c38df65caeaa1512c4 : SUCCESS in 18m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/aee117d60f75448c82cc8e23eff9f1fa : SUCCESS in 3m 45s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/e7a3b3730dbb427da0bf192a9f860a47 : POST_FAILURE in 40m 26s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/eca3a4d65a374fed81a1243a0ce85d38 : SUCCESS in 20m 07s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/7aee405256b0424da080013ae6ff3001 : SUCCESS in 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1efda83c2803f6eee36243fc1f51fc36d5552690","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-22 10:38:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b857c2982686f87c346ba704e7d34014690b1349","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:09:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84e6df61942a4aae956ef6ccd6d37a9a : SUCCESS in 7m 12s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d9be6b36a69c41a3b10477e756236c4f : POST_FAILURE in 21m 55s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/ffeb1977629149fc9afe91f37c8a5e05 : POST_FAILURE in 10m 10s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40d418e13b5f4dfa8420603ceee64e09 : SUCCESS in 20m 04s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9075d4abd9dc4c2d82749a7e505f4f02 : SUCCESS in 19m 44s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dcf9d92f57614d6f906b9af1cf42a2fc : SUCCESS in 4m 28s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/d569a5bea6004bc7870dbeea9811ea0f : POST_FAILURE in 22m 33s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/33d22c2439874f408de34392da3a9889 : SUCCESS in 18m 29s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/4884e5b5d8254a07a7803f3f617028b1 : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":2},{"id":"400eab620d082fdc9abf42eeb5351dba06f3f607","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-26 20:01:31.000000000","message":"Patch Set 2:\n\nrecheck - fix for POST_FAILURE merged","accounts_in_message":[],"_revision_number":2},{"id":"f382cae59d2f3ace2cb1c2bb85fc23ba428639c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 20:31:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7fa48fcfbec74098bf93663f537e5c02 : SUCCESS in 4m 58s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0c9b8c091ee843d78eb2d9bd2b6f9f82 : SUCCESS in 24m 34s\n- openstack-ansible-functional-opensuse-150 https://zuul.opendev.org/t/openstack/build/c9fd9ad0c76a47a385e4e0883eba6a18 : RETRY_LIMIT in 3m 14s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1a3e7a7d2d94725a768814f1c802012 : SUCCESS in 22m 33s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6424e4a210f44ab99968fd83336f90e3 : SUCCESS in 17m 26s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b6a93a72d93e44799bbbba76510aef06 : SUCCESS in 4m 50s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/8052cda954774c9fbacd847ad4380ef4 : SUCCESS in 22m 11s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/3b4bd3e995c04469a4ee391b4f1f77ea : SUCCESS in 19m 49s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/91ab46f846c541c08f0f4b4de68f4d41 : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0fb2c1787d700602ac0ff7b8357cf722c5e9109c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:36:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0db4aeac57cd230a39621ac2540189eb2aa09864","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-27 12:36:12.000000000","message":"Removed Workflow+1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"26b0b7669bfda65bf97406a0cdcb8811d9e6807e","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-27 13:01:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"008ebb3438280bf1cce57bfb1323d9f7865ec99c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:01:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"74bbe23f7a7ddb5bd0f6b23df5a049c58ce30422","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:31:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e4102775020f0ca5f80344f43ec0b6058afae31a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 13:31:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1fd4a37ac8a458599c142765f77fb11 : SUCCESS in 5m 34s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1dbe1624734f49e5859e3146dbf6bcb7 : SUCCESS in 19m 06s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15c705ab346e4649aaa9df69560af620 : SUCCESS in 22m 08s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/61ecf003d21c41e6adce7be89beab638 : SUCCESS in 17m 51s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5fe93d89d0234c78b391d5be8b2bafd1 : SUCCESS in 5m 09s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/4fff72f30ce143c9a108c26d818cc59f : SUCCESS in 22m 15s\n- openstack-ansible-functional-ubuntu-xenial-ssl https://zuul.opendev.org/t/openstack/build/3fe73d71a0784f969fe040b48baf92fe : SUCCESS in 20m 38s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/d64483383cc445d285b227ad6dd65ac4 : SUCCESS in 18m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"508ab4b90fa7fcc82d2d554ece61970e5eedde81","revisions":{"282f57ea2b4898cc7d04cfffae05121f049f5100":{"kind":"REWORK","_number":1,"created":"2019-08-19 07:46:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/32/677132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/32/677132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/1"}}},"commit":{"parents":[{"commit":"f551e6a51d6b4e80bca95e2dd030103883ce438a","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f551e6a51d6b4e80bca95e2dd030103883ce438a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 07:44:39.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-19 07:46:28.000000000","tz":0},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nChange-Id: I639f869084e71575b6db7a69fe9d06f67b25cfc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/282f57ea2b4898cc7d04cfffae05121f049f5100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/282f57ea2b4898cc7d04cfffae05121f049f5100"}]},"branch":"refs/heads/stable/rocky"},"508ab4b90fa7fcc82d2d554ece61970e5eedde81":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-22 10:38:28.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/32/677132/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/32/677132/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/32/677132/2"}}},"commit":{"parents":[{"commit":"f551e6a51d6b4e80bca95e2dd030103883ce438a","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/f551e6a51d6b4e80bca95e2dd030103883ce438a"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 07:44:39.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-22 10:38:25.000000000","tz":0},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove the job, we test with openSUSE 15 now (as\non master). This allows to remove the nodes from CI.\n\nDepends-On: https://review.opendev.org/677947\nChange-Id: I639f869084e71575b6db7a69fe9d06f67b25cfc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/508ab4b90fa7fcc82d2d554ece61970e5eedde81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/508ab4b90fa7fcc82d2d554ece61970e5eedde81"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
