)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~687454","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Id15d75c8339b253b516fe5d7ce7fc95a812237ef","project":"openstack/openstack-ansible-galera_server","branch":"master","hashtags":[],"change_id":"Id15d75c8339b253b516fe5d7ce7fc95a812237ef","subject":"Bump galera version to 10.3.18","status":"MERGED","created":"2019-10-09 06:10:23.000000000","updated":"2019-10-10 16:55:05.000000000","submitted":"2019-10-10 16:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687454-1570726439053-0f5a3271","meta_rev_id":"c6a83ddcf5ece8bb0b651a40db3e186ec7af002f","_number":687454,"virtual_id_number":687454,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 16:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-10-09 18:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-10-10 14:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-10-10 14:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 18:45:06.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-10-10 14:48:44.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-10-10 16:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f514a15c06d9db973bf171d69fdd176078b178f3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-09 06:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8c1d8642458eb992a229969d75cb1313d94ca43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 08:21:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/971c7972aa514057bf93a555c02b2376 : SUCCESS in 5m 48s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/a767b20c0a5240d89867b34835c15a67 : SUCCESS in 15m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/657cac75cbd443b098643519a951827c : SUCCESS in 18m 28s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05adaadb12254a95953c82e10da8e957 : SUCCESS in 19m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/95ed19545ca64a69b90f31f989c322c0 : SUCCESS in 1h 46m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a075f0dfba8244b0ae44997e16507fc2 : SUCCESS in 2h 04m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cd1f1b85fbfa4d658c58d0677d948e1b : SUCCESS in 1h 23m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fb15301658640e8bb36bcf68a66acff : SUCCESS in 1h 01m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d69be287fe9843d1b09916b69c83da37 : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6df94923c3d74109955f9e9da299150d : FAILURE in 44m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/086f7ba42a244f588013423a3967b690 : SUCCESS in 46m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10238915dad747f691133d72940944a9 : SUCCESS in 3m 16s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/ae11a3f478c44ba79e4ba9e8a554f775 : SUCCESS in 21m 28s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/93aa230330bb4fb48f37aa0a351ef8c8 : FAILURE in 31m 04s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/70b3d7c62b6841d5971af7ebb88c7dcc : SUCCESS in 16m 55s","accounts_in_message":[],"_revision_number":1},{"id":"cab1f80dea2e9f3a4e29743b0e81710d82331d53","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-10-09 18:45:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f7d33ff3a4fb556f6438df258c40c33c502423ab","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-10-10 14:48:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2386c412e02f65dacf621134262612db080a8fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:48:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1e2126456f725ae5564823ef271466c7665acd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:53:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b17c55b284c9441d89c0fc5d1e53643f : SUCCESS in 5m 02s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/2b085c0284374df0b877b9f99ab4dbf3 : SUCCESS in 17m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/86a82d2c027b41c6b29094241e18016a : SUCCESS in 24m 57s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d553a33bdf894db18adb7f756ea138fc : SUCCESS in 25m 43s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0a4a3947b74d4615bb7f1f03f0baaa0b : SUCCESS in 1h 20m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9975a60ff26e46f7a3425b954acb67ae : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dca0def8a3e74678a26133054c0d8559 : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4a632314198428cb6d977ddaef4e0d7 : SUCCESS in 2h 02m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/db1c9c42b4c9474087354a9273b00694 : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9e273d41d5b48cfade59b4c0e917957 : SUCCESS in 47m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38452003acbf43ff85010c674848c9ee : SUCCESS in 2m 53s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/7247192ba6b24b0c9111d7830f88e21d : SUCCESS in 21m 25s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/f0f2412b54fe4543aff0454b0bf84d12 : SUCCESS in 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1c3bc13fdf970b27cfd40fafa54302a28a4afc77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:53:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6a83ddcf5ece8bb0b651a40db3e186ec7af002f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 16:55:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fcb921e2c2444c79007e7e2e5145131 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73ce6a8c9a226c1ff96bf1a7afed6dbaf976c67a","revisions":{"73ce6a8c9a226c1ff96bf1a7afed6dbaf976c67a":{"kind":"REWORK","_number":1,"created":"2019-10-09 06:10:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/54/687454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/54/687454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/54/687454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/54/687454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/54/687454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/54/687454/1"}}},"commit":{"parents":[{"commit":"aa6bb9bf9b88f268e737fa71db363898cded4dcb","subject":"Move to PyMySQL instead of MySQLDB for ansible-2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/aa6bb9bf9b88f268e737fa71db363898cded4dcb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-09-12 14:11:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-10-09 06:09:01.000000000","tz":60},"subject":"Bump galera version to 10.3.18","message":"Bump galera version to 10.3.18\n\nChange-Id: Id15d75c8339b253b516fe5d7ce7fc95a812237ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/73ce6a8c9a226c1ff96bf1a7afed6dbaf976c67a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/73ce6a8c9a226c1ff96bf1a7afed6dbaf976c67a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
