)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~688800","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I0ea6477cb62773a1c9c82e77204c9bc72255a892","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"I0ea6477cb62773a1c9c82e77204c9bc72255a892","subject":"PDF Documentation Build tox target","status":"MERGED","created":"2019-10-15 19:54:30.000000000","updated":"2019-10-17 14:06:46.000000000","submitted":"2019-10-17 14:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688800-1571321104809-d762e54f","meta_rev_id":"af482146045319c56520f002b51664303ea775a9","_number":688800,"virtual_id_number":688800,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 14:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-10-17 13:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-16 20:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2019-10-17 13:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 20:28:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-17 13:38:16.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-10-17 14:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbb495f2949dbc99dacdc55e0e23a11085c12801","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-15 19:54:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a644a2be4eea4a1082852507dbcdac67a97c7927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:29:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a6788ea969ad4e53ad0db44867a74a82 : SUCCESS in 5m 32s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/06d168ed80054901908a287fdfa981cd : SUCCESS in 16m 49s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/3dc608679a334206a6ed59a2cbfe2fea : SUCCESS in 20m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84e5546d2d8a4be2b78627a2d84aa0e5 : SUCCESS in 18m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/efc60367ed9b405f819db82570eae6a6 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518140c921644592aa9bc439119de2b9 : FAILURE in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fd82f0dd43c41a580307f52364d50cd : SUCCESS in 3m 03s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/be3665f9a6df49e3b4cec1166715b384 : SUCCESS in 20m 18s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/88dd2d263da24231a717a2f3d8d234e7 : FAILURE in 29m 51s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/a8a478cb3fa54f26a6b6a8242a9e8bfb : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e7f728f963ed78134d4523aeea819a50c4fc4284","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-16 11:15:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"dd669f60e1e7768ed7d57d39dd4808cfd6c4a2dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-16 11:26:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b393e475308058c97fdb86796a99efe1b08e0a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 12:01:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/12ceb12357834567b881527901af4bfa : SUCCESS in 7m 45s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/bd68d9cc63df4abebf8b81c3967bde6d : SUCCESS in 17m 07s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/86471814c1254a5aba64228d6bdc32c9 : SUCCESS in 20m 00s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9af5304226c142eab4e720a72888dd4c : SUCCESS in 18m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/441ee1b6369b4773aee6f05265bfca70 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e085343e2842eab911d67c627bf667 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f3a5adf9ff14158a2ab2f82ebc9cd84 : SUCCESS in 4m 27s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/b19aad35b6db40f2916f71ebebda4589 : SUCCESS in 24m 23s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/5e45572d495a4dc7a9f738724599cba6 : SUCCESS in 31m 58s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/bd9067b5855a4373b539f2f3e6be3cb4 : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":3},{"id":"fd235b70d0ff09b4efd4cecb2637626dd234fdb2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-16 20:28:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ffbf89eb323cec8698d530c5bb0b975517aa686f","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-10-17 13:38:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fb1ecc779a97c62269fd4dad785321dd6a11a40e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 13:38:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"19e5c0dd91414e9563059f1ed7ac3d108cb455fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 14:05:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c1ff9df63d28a141e6a76ff71295cb30af6854fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 14:05:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/70434bd2dfa145588c9890d5c3301786 : SUCCESS in 7m 16s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/40bc7af935604fe385adf78e633537ad : SUCCESS in 15m 50s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/e2a5ba2ecade4e62bc2c0023d4fa8e72 : SUCCESS in 21m 16s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b650dee8f584b599368fe83c6b1e723 : SUCCESS in 19m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92990abf9f9a4f6c8c91e92e372b6435 : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9ed097daf74c4fb2b31f0630913a21 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b42ba59a06c47f298ab226f3ec02f19 : SUCCESS in 4m 08s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/9f98d9f04a584aa49e073ce9d815fe9a : SUCCESS in 24m 13s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/7aaa0913182c4caa918b28dc1c91e22a : SUCCESS in 17m 56s","accounts_in_message":[],"_revision_number":3},{"id":"d35f84a38ee9ab91295ce3c30de93ac722e61aa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 14:06:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c729b80ca77342f1a7c56e27295bd66c : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da22dbc6ea5f436a8d77a55a392c7b3e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7132d760bee431635855559596803a6f80aa3090","revisions":{"642f9b720f17f396ffe42b672ddef9399e2c3535":{"kind":"REWORK","_number":1,"created":"2019-10-15 19:54:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/688800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/00/688800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/1"}}},"commit":{"parents":[{"commit":"60880f7d0df5ae580f216f356837590f7931dcd7","subject":"Merge \"Bump galera version to 10.3.18\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60880f7d0df5ae580f216f356837590f7931dcd7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-15 19:53:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-15 19:53:45.000000000","tz":180},"subject":"PDF Documentation Build tox target","message":"PDF Documentation Build tox target\n\nThis patch adds a `pdf-docs` tox target that will build\nPDF versions of our docs. As per the Train community goal:\n\n  https://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nAdd sphinxcontrib-svg2pdfconverter to doc/requirements.txt\nto convert our SVGs.\n\nChange-Id: I0ea6477cb62773a1c9c82e77204c9bc72255a892\nStory: 2006105\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/642f9b720f17f396ffe42b672ddef9399e2c3535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/642f9b720f17f396ffe42b672ddef9399e2c3535"}]},"branch":"refs/heads/master"},"556073d81684def0c33ab77581a816a4c20721f8":{"kind":"REWORK","_number":2,"created":"2019-10-16 11:15:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/688800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/00/688800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/2"}}},"commit":{"parents":[{"commit":"60880f7d0df5ae580f216f356837590f7931dcd7","subject":"Merge \"Bump galera version to 10.3.18\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60880f7d0df5ae580f216f356837590f7931dcd7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-15 19:53:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-10-16 11:15:05.000000000","tz":0},"subject":"PDF Documentation Build tox target","message":"PDF Documentation Build tox target\n\nThis patch adds a `pdf-docs` tox target that will build\nPDF versions of our docs. As per the Train community goal:\n\n  https://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nAdd sphinxcontrib-svg2pdfconverter to doc/requirements.txt\nto convert our SVGs.\n\nChange-Id: I0ea6477cb62773a1c9c82e77204c9bc72255a892\nStory: 2006105\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/556073d81684def0c33ab77581a816a4c20721f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/556073d81684def0c33ab77581a816a4c20721f8"}]},"branch":"refs/heads/master"},"7132d760bee431635855559596803a6f80aa3090":{"kind":"REWORK","_number":3,"created":"2019-10-16 11:26:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/688800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/00/688800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/00/688800/3"}}},"commit":{"parents":[{"commit":"60880f7d0df5ae580f216f356837590f7931dcd7","subject":"Merge \"Bump galera version to 10.3.18\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/60880f7d0df5ae580f216f356837590f7931dcd7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-10-15 19:53:45.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-10-16 11:26:26.000000000","tz":0},"subject":"PDF Documentation Build tox target","message":"PDF Documentation Build tox target\n\nThis patch adds a `pdf-docs` tox target that will build\nPDF versions of our docs. As per the Train community goal:\n\n  https://governance.openstack.org/tc/goals/selected/train/pdf-doc-generation.html\n\nAdd sphinxcontrib-svg2pdfconverter to doc/requirements.txt\nto convert our SVGs.\n\nChange-Id: I0ea6477cb62773a1c9c82e77204c9bc72255a892\nStory: 2006105\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7132d760bee431635855559596803a6f80aa3090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/7132d760bee431635855559596803a6f80aa3090"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
