)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~709468","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~I27b137dd7bb3dc8728ff0be47dcbce549a0e0be4","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I27b137dd7bb3dc8728ff0be47dcbce549a0e0be4","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:05:44.000000000","updated":"2020-02-25 13:47:09.000000000","submitted":"2020-02-25 13:43:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709468-1582638229277-2a34db55","meta_rev_id":"dc219e8bd9c9ff8573f22087bde6995adf75ff8f","_number":709468,"virtual_id_number":709468,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 13:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-25 11:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-25 11:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 11:55:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-25 13:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20efae3c51b05a16ad5ab8f69de2d649661ca50c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:05:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3da0f66704f87ad42f584295a506450a97150e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:41:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9a9f006ec5b840dc8e838bd6f2233899 : SUCCESS in 8m 46s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/a546427ce64e4bd4ae69358b07331452 : SUCCESS in 17m 12s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4ab7e671b5df499fbf2d1279a310f724 : SUCCESS in 20m 38s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bbb3274fa11449387fc3befe49f8803 : SUCCESS in 19m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/91a0ec373f5c48fbb662165292929578 : SUCCESS in 1h 06m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/396cbd615c0c47cdb7121c24cc24db6c : SUCCESS in 1h 33m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/913d2c8ad5f74a80afd13c85192fe93e : SUCCESS in 1h 25m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeed73bddae248f483de1e7c51455f03 : SUCCESS in 1h 09m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/be89efe6d4474a53b36746bcedba5087 : FAILURE in 1h 27m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6b14b73e74a4fdda4edc394f0b2b2d2 : FAILURE in 59m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/968dd33cd16d4b36b96452586f621e6a : SUCCESS in 1h 16m 28s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/40c15e218c4749349309bb8279d01b5d : FAILURE in 30m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54ba99d7f7e34833a806b804fbcfc8ce : SUCCESS in 54m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15bac319e5924475bdf04ae2e90b862d : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51dd006cdf9c45fb8eefe1c39397b23a : SUCCESS in 5m 44s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/2e30182781cb41189cdcde81f7e23179 : SUCCESS in 20m 31s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/ea0d77dd033841e6b8809bcef4491595 : FAILURE in 7m 47s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/c021507db7b74557be83e1c0b54fdbff : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7671a44ca00208852cc8a093d06441df3cd48c28","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:53:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"301118ce0bc44eb73fb8c547f5b4723f40cb5759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9b70a5ec8f5b471086d1a27f0deb65df : SUCCESS in 9m 36s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/1b814314a0974284afbb53ffdc20e42b : SUCCESS in 20m 39s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1a877b4ca0a249868d6eea71e89c3a96 : SUCCESS in 25m 51s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe7b61bc9056442d964fb819bd7bf755 : SUCCESS in 18m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f6b6bc8994149649fed64645c9b7f5f : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b3c3d1c721f4e6580bac726d524a859 : SUCCESS in 1h 41m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/37ebfc575762412580d5b3b612485ef6 : SUCCESS in 1h 32m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5be3245101b643798ca1a61f64a1d71d : SUCCESS in 1h 17m 40s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a2e4c41929d7464ea765eef9c4749a43 : FAILURE in 1h 25m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65b0eca0cdbe4248a2b168c7f0a18a32 : FAILURE in 1h 03m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/554d00b0a2684b85a2fd3dd04245b96b : SUCCESS in 1h 10m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/abd98804c06649e697a02c9ef507100b : FAILURE in 32m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1de2872a27314bbd9917121ed1a77393 : SUCCESS in 1h 05m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82122e3c1b1247c4b3aba4ba1673523d : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77eaf393d8b04a37b941d7cbe508d63e : SUCCESS in 6m 08s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/5b9db7e877ef413d97f45615b028a559 : SUCCESS in 23m 51s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/e383ed84d06c455fb80e8e0eb089cbe3 : FAILURE in 5m 55s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/cf1b2d27b6a0469dadea245ca1b680e9 : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":1},{"id":"04334d60b7d792190dbce2a14665e68b4c56c31c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 11:55:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1d433099b1b777cc9b6dec25b9e380bfc602103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:55:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e68d228fe652d5c3c4d2717f3942bd19da5a0238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:43:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/91aa952df3784ddabc3a58fbd651c5fc : SUCCESS in 8m 14s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/66e4b0902bd4403aafd23d787c679c81 : SUCCESS in 18m 33s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4a231932f5d24431ae942032b0853aaa : SUCCESS in 22m 29s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/419b35d7741548caacf37aa073353043 : SUCCESS in 22m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e6f893b1d03f41feaac3f1367b195347 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7fc795c354e44216856eeee47a299890 : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1b6bb36656ce487182416063a55fa56a : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79c1f15a2a7c489d837c9a81f7736ded : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/17861d923c42404ab15de217aca0e102 : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d21d8aeaf19b40b49eb2556c56e0a29c : SUCCESS in 54m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a276ff7e6e704cb8b7110fb2e90df321 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2acf3cc02f848e2bf9164dad67cb040 : SUCCESS in 5m 15s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/4c2ee02adbc349bc8752cd7083a78f21 : SUCCESS in 26m 23s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/6867ea62493f45d7a4f43690855bce8f : SUCCESS in 21m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5f8dd2e6bfbee793f7f8c3352a5048c107825cfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:43:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d26f4e5569788c2ef93aa617184da3de2285317","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:47:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5faf370891e4440f8c88f8b6758b780d : SUCCESS in 2m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/241397bd65264ad59c4cc4532bb8e88d : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"299fb6e106fb85a7e068be34c6b7cd1468ad63f5","revisions":{"299fb6e106fb85a7e068be34c6b7cd1468ad63f5":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:05:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/68/709468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/68/709468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/68/709468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/68/709468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/68/709468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/68/709468/1"}}},"commit":{"parents":[{"commit":"ec4f967b1386f4b83c1de2baa9da8ba37b00823a","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/ec4f967b1386f4b83c1de2baa9da8ba37b00823a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:05:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:05:39.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I27b137dd7bb3dc8728ff0be47dcbce549a0e0be4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/299fb6e106fb85a7e068be34c6b7cd1468ad63f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/299fb6e106fb85a7e068be34c6b7cd1468ad63f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
