)]}'
{"id":"openstack%2Fopenstack-ansible-galera_server~711689","triplet_id":"openstack%2Fopenstack-ansible-galera_server~master~Id857c6d36ff371b4e4dbaae19fff80a6d3bec9c0","project":"openstack/openstack-ansible-galera_server","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Id857c6d36ff371b4e4dbaae19fff80a6d3bec9c0","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-03-06 17:53:57.000000000","updated":"2020-03-06 22:11:56.000000000","submitted":"2020-03-06 22:10:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711689-1583532619059-c6fdaf72","meta_rev_id":"9d2377688ae9f6d2e1049f9d0a403bf349922287","_number":711689,"virtual_id_number":711689,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 22:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-06 20:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-06 20:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 20:39:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-06 22:10:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3305166599dbf4884ad3b03aea0f9b0abbf5feab","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-06 17:53:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6f792201e5b9dcc9ad14c38d6c31e78fa7fb046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 20:29:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7a6b27e695048c1b72c1f2ff5154f4e : SUCCESS in 7m 32s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/c26318a9f5e3469a8ac2706f62efa73d : SUCCESS in 20m 25s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/69471b05defd490ca102838179a25670 : SUCCESS in 23m 31s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3922578fd844667ba7f00e23e8a30c5 : SUCCESS in 20m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/011eb815ec3c469fb9bb66c708536590 : SUCCESS in 1h 07m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d0882f48e7a4f29a72ab01e4a084a9a : SUCCESS in 1h 28m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3e340cf09c25425ca3ee154287855444 : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/113680146de64f45ad94a0ca84e3ee16 : SUCCESS in 1h 12m 28s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e455a2060aa14a0a8cbdbea012408a0b : FAILURE in 2h 30m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a33860ab21f4bf98276f5706b5378ef : FAILURE in 1h 04m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cbe2c0fee033473bab655405a28235dd : SUCCESS in 1h 19m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e18c57a0f8254944a2e0d1d2f328bb44 : FAILURE in 31m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/515f479f9f3a4c56968a7ef1651354ab : SUCCESS in 54m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ee8577299b4f5d9bdecda8a3724fea : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a53544c96d24bec8193ddd836044e49 : SUCCESS in 4m 11s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/e91c781a5159422186cc389f5d083a79 : SUCCESS in 22m 10s\n- openstack-ansible-functional-opensuse-15-ssl https://zuul.opendev.org/t/openstack/build/8ebfe8182e9448759e255bcd7a09c1de : FAILURE in 6m 52s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/8a6102e678f94788a89cfe808bcf8835 : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3d657e45a962bfb0e58ab8d193001e041d57a1a8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-06 20:39:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5db4042f502cb54752a68de7c9c613cf1b4085b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 20:39:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff66ebb1c582f36abd01879effaaf98f70a4c94e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 22:10:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8c6f9aacbfe1453fb1f1a250e398c4a9 : SUCCESS in 7m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9b00cbb71e9b459391b5f1c96eadab32 : SUCCESS in 23m 52s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0078454eee9c478b9f947fb80635dee3 : SUCCESS in 22m 19s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/599fdc8c5ed24408b629d86afd73a67f : SUCCESS in 24m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/131e2bd3049048308ac7802f576f26bb : SUCCESS in 1h 02m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/217a47d2dcb14f95ba110d92124fd99f : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3d076ec6d7a746d9851de8aa80f28aaa : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af321bca12a3465f8e57b648e172299f : SUCCESS in 1h 10m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/90852d3b30154b6c860ef2a0f37f6991 : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2997cebf5ead43b59783282d83bee81e : SUCCESS in 54m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb4e6cb01e94044aceef4aad9b1dc69 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b485c4b3b8d24522a54806b8b76dd791 : SUCCESS in 4m 06s\n- openstack-ansible-functional-centos-7-ssl https://zuul.opendev.org/t/openstack/build/f90e35c5726e4faca243e4ab9d915cd6 : SUCCESS in 22m 54s\n- openstack-ansible-functional-ubuntu-bionic-ssl https://zuul.opendev.org/t/openstack/build/af592c95fe844bf384a00f947bc8f2d1 : SUCCESS in 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"030db8f9aff38a5cbef097658f13b24b5425d646","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 22:10:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4781c413d9a2324330c5571b448d381371c9523e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 22:11:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3dfda177c704d7a817869caeb7bd47a : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b66df1a48724af9a307fef3fbe6e05e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07d8650b517ad2302373a7ba16ea1b2ff70e3ff7","revisions":{"07d8650b517ad2302373a7ba16ea1b2ff70e3ff7":{"kind":"REWORK","_number":1,"created":"2020-03-06 17:53:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/711689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-galera_server","ref":"refs/changes/89/711689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/89/711689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/89/711689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/89/711689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-galera_server refs/changes/89/711689/1"}}},"commit":{"parents":[{"commit":"e3f31bea6e8e318e5860deee100c0890d8e5506b","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/e3f31bea6e8e318e5860deee100c0890d8e5506b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:53:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-06 17:53:55.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Id857c6d36ff371b4e4dbaae19fff80a6d3bec9c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/07d8650b517ad2302373a7ba16ea1b2ff70e3ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-galera_server/commit/07d8650b517ad2302373a7ba16ea1b2ff70e3ff7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
